)]}'
{"id":"openstack%2Fpuppet-oslo~805510","triplet_id":"openstack%2Fpuppet-oslo~stable%2Fussuri~Iba93ffdacf45096a6db65f3f0dfcfaf7d87574f2","project":"openstack/puppet-oslo","branch":"stable/ussuri","topic":"idle_timeout-stable/victoria-stable/ussuri","hashtags":[],"change_id":"Iba93ffdacf45096a6db65f3f0dfcfaf7d87574f2","subject":"Fix the wrong default value of idle_timeout","status":"MERGED","created":"2021-08-21 18:39:28.000000000","updated":"2021-08-24 08:59:26.000000000","submitted":"2021-08-24 08:59:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805510-idle_timeout-stable/victoria-stable/ussuri","cherry_pick_of_change":777127,"cherry_pick_of_patch_set":1,"meta_rev_id":"3696267fcf062b607e376d280071dd7ce835b1b2","_number":805510,"virtual_id_number":805510,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-24 08:59:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-08-23 14:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-08-24 06:56:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-08-24 06:56:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-21 20:00:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-23 14:43:15.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-08-24 06:56:56.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"798ccb3c517c70689af2aeec581a9bd8e81db24a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-21 18:39:28.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"9bcc8cacf96f25d9c3e020fe65d6e91535f85732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 20:00:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e0560ed42c34e0e96626e31c34418e9 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77be7911619a4276bc62ac25c82cd8ce : SUCCESS in 3m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0cb8ca52d7a4feabfdf1d23e9296c14 : SUCCESS in 3m 25s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7fb6702b2dc42d3b6e1cb9e1a804134 : SUCCESS in 9m 00s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/b3048961cb3d446593cc27e26a7a5de9 : SUCCESS in 8m 58s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/053d2b7c52b44c3f8b8b16e798e95bda : SUCCESS in 7m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b81f4088bcd740ed88f2a07fdaa24d9b : SUCCESS in 6m 56s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9ff9b41632fe4ac58d70165d381fc646 : SUCCESS in 51m 45s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4791f6bd4ac476e8e8b5731cc460134 : SUCCESS in 44m 19s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/211f7f2ea05f4f58813470e732600f3d : SUCCESS in 39m 19s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3ddaa3a8704b4a57a81a1f3fd4eb5c9e : SUCCESS in 40m 48s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/12398c37b0714c8087a3e5722a7b4b5f : SUCCESS in 56m 48s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/077334a764ad4e27b2a86ec4355d5f27 : SUCCESS in 1h 02m 57s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/08d737e6126f497f8d3a71953326d474 : SUCCESS in 1h 07m 01s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1ba269fd632645a7a487605900b88a4f : SUCCESS in 47m 38s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/50ec5df2584d443c92d4bff6ef9756ed : SUCCESS in 58m 18s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3a369d363364e67ae7bb57d663b8c80 : SUCCESS in 1h 04m 22s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6caf6e02d61b4c8797a17a066adc17ee : SUCCESS in 1h 02m 50s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b0abaee04bd48c1b066ad409585f39b : SUCCESS in 47m 39s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/a7f466350eda489a966eba8a76d8cca8 : SUCCESS in 8m 44s (non-voting)\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/c03542fa88e74623b473a7dcf998b240 : SUCCESS in 8m 17s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1488142241684ce5a274be9bb490c984 : SUCCESS in 6m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aca3319b66ee7b59202483d70a8526f7ed069849","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-08-23 14:43:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3816c80b38abb570c24ac73a884c7bce48f733b7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-08-24 06:56:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c79c64232e722aa0f05b64c24b025c71b1ffb75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 06:57:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc3dbfc95102b01655722112d45570491f50dc3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 08:59:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b01707d27e4749d79dbe9eede80c1062 : SUCCESS in 6m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e367a8cd8f754065ab6d8cea71381dd4 : SUCCESS in 4m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0f8acd8f94e4a4085280fd964f8edf6 : SUCCESS in 4m 42s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/9bfc90b445cf4a65be80bdf315fe5a39 : SUCCESS in 11m 47s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/af482b0f2ffa4887916054aebb9a5ca0 : SUCCESS in 11m 53s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc335b4e53454f2598132f3ab1591ed5 : SUCCESS in 12m 01s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e83627759985442eb174b5d7b78d54d4 : SUCCESS in 56m 51s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/012cea7230dc4ae9be6d9b0809084c80 : SUCCESS in 1h 13m 28s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b51e1913ef3a4d0bb174a76ae9ca9b0f : SUCCESS in 1h 16m 20s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b46909357fcf4869814fe99209ad7fad : SUCCESS in 48m 56s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/9ac295d9dd874f30ba844ab06da9ba83 : SUCCESS in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3696267fcf062b607e376d280071dd7ce835b1b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 08:59:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06766e8789f411e6e135f041a6b264bb70d48ab8","revisions":{"06766e8789f411e6e135f041a6b264bb70d48ab8":{"kind":"REWORK","_number":1,"created":"2021-08-21 18:39:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/805510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/10/805510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/805510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/805510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/805510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/10/805510/1"}}},"commit":{"parents":[{"commit":"a5869c721e2510519c1beefb6b2efb15cbc373db","subject":"Merge \"Cors: Accept array for ListOpts\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/a5869c721e2510519c1beefb6b2efb15cbc373db"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-15 00:44:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-21 18:39:28.000000000","tz":0},"subject":"Fix the wrong default value of idle_timeout","message":"Fix the wrong default value of idle_timeout\n\nCurrently we always get the following warnings because of the wrong\ndefault value of idle_timeout.\n\nThe idle_timeout parameter is deprecated.\nPlease use connection_recycle_time instead.\n\nThis change fixes the default and ensures that the warning message\nis shown only when the deprecated parameter is set.\n\nChange-Id: Iba93ffdacf45096a6db65f3f0dfcfaf7d87574f2\n(cherry picked from commit 12e2f4a886f7a97d1f34bce22254ea114f829bbd)\n(cherry picked from commit 69263d76c1e9d925b96a36beba521dde756863e8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/06766e8789f411e6e135f041a6b264bb70d48ab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/06766e8789f411e6e135f041a6b264bb70d48ab8"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
