)]}'
{"id":"openstack%2Fpuppet-oslo~806485","triplet_id":"openstack%2Fpuppet-oslo~master~I34f10b51e1179beed327b9544b0899105cb30895","project":"openstack/puppet-oslo","branch":"master","topic":"cache-package_ensure","hashtags":[],"change_id":"I34f10b51e1179beed327b9544b0899105cb30895","subject":"cache: Allow customizing status of the backend package","status":"MERGED","created":"2021-08-29 02:32:21.000000000","updated":"2021-09-20 11:13:16.000000000","submitted":"2021-09-20 11:13:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806485-cache-package_ensure","meta_rev_id":"3f354c7c24f5215fff5ed79312974fe134b3c954","_number":806485,"virtual_id_number":806485,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-20 11:13:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-09-02 16:13:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-09-20 06:04:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-09-20 06:04:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-29 04:50:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-02 16:13:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-09-20 06:04:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"8acc20cce453bb8911c00f319d9aaf425334a7a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-29 02:32:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82e87aea3de43f66103365ada602a03439315280","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 04:50:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f070a310492c424a88ee67be05c2bbb8 : SUCCESS in 7m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec6c602e296043c9994c8232931c1b3b : SUCCESS in 4m 08s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b95752cc53344f70b505b04a04d99aea : SUCCESS in 4m 29s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/88d2676c09a243b1b21f60555377d078 : FAILURE in 15m 01s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e2e3c259c9e46bcb0c7724fbabab5a4 : FAILURE in 21m 02s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03c9d989d2854414b41c3e837e727c2e : FAILURE in 17m 44s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffc7e5de5ddf412b9e36bc7f77d0bf9d : FAILURE in 11m 41s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82e0599f95bd4f9d9d31636efb65e334 : FAILURE in 11m 03s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d69ec75773e54bb88efbb315b3c5d499 : FAILURE in 1h 06m 35s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a69f700321724ba7b2cb18f3e322400c : SUCCESS in 1h 02m 32s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/975d3cff8e744d09a36b91e5f1023bbf : FAILURE in 30m 20s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c830ad083a84d0cbfd1a6e33fc4eb64 : SUCCESS in 41m 15s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d7f78dc978a49f29d3f79930e2b6406 : SUCCESS in 25m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/94c894665045402296af693547f30e85 : SUCCESS in 1h 08m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b0445a8ed1e41ecb526da160b5134bf : SUCCESS in 1h 56m 12s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cca62ebef46c44a28cb61f615d2e281b : SUCCESS in 1h 16m 59s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/38b9e2fa592f4c63a8507ceac51a5c79 : SUCCESS in 54m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/606cc70e7964411fa70f0788a551d880 : SUCCESS in 4m 49s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4ff67b6524c9422ab87088c6a5e8f106 : SUCCESS in 2h 15m 51s","accounts_in_message":[],"_revision_number":1},{"id":"cd6afe420c753e75028331a6628326447a4bf3a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-29 05:31:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2afcaa5feb1d525d4a111323e2499be9512eacf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-29 05:34:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"afeb07f42c61b08727cb13d5434f5c37257f85cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-29 06:06:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"452bda84186e1692457fe7f4711bbe2abd9e1f19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 08:17:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0946748a50244d0587a20faeb4c93f73 : SUCCESS in 6m 37s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4861102e3a8a4012b75bf49e5e071fdc : SUCCESS in 5m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18bac9c44a1140d68217261df1ac74f4 : SUCCESS in 5m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a29558b848c496a9e667cde54b234c0 : SUCCESS in 14m 49s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/8223a81de5124c928c94184cee820517 : SUCCESS in 13m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/989add60e1dd4ff5b9baf32d36a47654 : SUCCESS in 13m 09s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a18e39bbf5a4c2196bec3c70839a387 : SUCCESS in 13m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f8b0bd42c2a4649a9860bc9329c374b : SUCCESS in 10m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70865b4a4f284a9fb3718d9452eb3b40 : SUCCESS in 54m 02s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa3c5a2b28c34f488a3ff5424d00805f : SUCCESS in 52m 27s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3133d92ef9614652977c7bfcd9794b34 : FAILURE in 34m 39s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb9b0f6e5fa34deaa3d1023c1fbc284a : SUCCESS in 39m 57s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5dbc13774f7d41d8b09ba3cb35b8b02d : SUCCESS in 29m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e5b09f79f8749ee89e10190acc9d93a : SUCCESS in 1h 10m 33s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f444edd480d4f3582b2759439260cd9 : SUCCESS in 1h 35m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/628de4c5c73e422fba809d0297e4eef2 : SUCCESS in 1h 31m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c596d3a7ceeb4cf5ba70be4027a46e5a : SUCCESS in 1h 07m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b908ab050b34990885df635b730fad2 : SUCCESS in 5m 19s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fddad8b1b2a8416098018679d609c53e : FAILURE in 1h 44m 44s","accounts_in_message":[],"_revision_number":4},{"id":"8440b563bbbf198355db57a0f5e52e3a426c665d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-29 08:33:32.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"827b61b7f756a342d609ac947ba2416d777c718f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 10:29:15.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0835e32925cf4588b30c84761b77ec52 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c91996f235442778cadc04a42642b5c : SUCCESS in 4m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7aa91239266545c2b85169eb80a21dc8 : SUCCESS in 4m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/52aba92c937746d9a15877a7ca6406f5 : SUCCESS in 10m 16s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b17562ad3e9a4b7f9593af060449cf0f : SUCCESS in 13m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1382027780574d34a60cfbcbff4e31e2 : SUCCESS in 8m 03s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/943ae1ef58e14fc5b2ed406be18b1c2d : SUCCESS in 11m 36s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47135f00836b4208981e4318dab4cf04 : SUCCESS in 8m 00s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebc776a7b210434bbbf4e4483d92f1d4 : SUCCESS in 57m 11s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c55fff2dbc5c4b33b24ac553675ea205 : SUCCESS in 56m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f0f6365230549a38c504e563a00eecd : FAILURE in 28m 42s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f42c9d18ed914a82be00600234b5d001 : SUCCESS in 34m 16s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7001f6e98fce49f3a7dd92994f11141e : SUCCESS in 22m 15s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/160bb903ad414daa8a4d89acbed94c2a : SUCCESS in 1h 26m 54s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c8d8a3b7832541108bacb9299d453e22 : SUCCESS in 1h 20m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1127e5207ef049dc94370779529d4d43 : SUCCESS in 1h 23m 19s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/39835fa0f13b476ab36564760acde8f6 : SUCCESS in 59m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f499490b2f24d08af344d529f7642e4 : SUCCESS in 7m 51s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7930cf86635d43488664fbcc4d6f5668 : FAILURE in 1h 51m 06s","accounts_in_message":[],"_revision_number":4},{"id":"114044db0c1067a1c4473baf1a2dc33ae93dc152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-30 08:01:13.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"4fadea145576a1005c0b3ac9029f2aee8ff415ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 10:03:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd5a0f37eb4d4f88b761a8954f0650d7 : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb29a63ae5e1480e95eb4207d9bbd8f2 : SUCCESS in 4m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9403b994f0f476f8f6095260cf20383 : SUCCESS in 4m 45s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c12c0b9aa97466e9950f8dbb46c7763 : SUCCESS in 11m 12s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/60e82c91ea4b48228dc2e234de2b14fb : SUCCESS in 15m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc68bc7225074ceb94510c3b72d5c15a : SUCCESS in 10m 10s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3098e3ce8ab347169ba82e7869592bc8 : SUCCESS in 9m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a25c7100cd904393b9bfbe5575b84957 : SUCCESS in 9m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a767c8741f984ed4a53784cab83382ad : SUCCESS in 51m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d5186f82eb44beeb9581d9b9182c325 : SUCCESS in 58m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1465a5fdfbfe4c17aec27a5d83d904f5 : FAILURE in 28m 03s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/881c4108307c499b8488da6584092c58 : SUCCESS in 40m 44s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dff937312532466eb1535d2cba882805 : SUCCESS in 23m 35s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/86310a53af3c47b5ba0fd4329d46ac31 : SUCCESS in 1h 17m 10s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c8020ad7e2348cf95b6fb4c44b273ff : SUCCESS in 1h 22m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/30a520f969e7402c8f60fec5ce18669e : SUCCESS in 1h 28m 35s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/03c2ef95ca494a8a8c5c93298e6f942c : SUCCESS in 1h 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c19a86374d2448b188bfd9e308a1389a : SUCCESS in 4m 58s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4160d043e6e14df6884dd088566fdeda : SUCCESS in 1h 57m 41s","accounts_in_message":[],"_revision_number":5},{"id":"8a59987157e564283f70d41177c1c5fc0763df63","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-02 16:13:21.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"029d988076a3f132284e756e8974f19645c39c6b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-09-20 06:04:23.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a30f9721bb1ac417ec620f940248885f85905987","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 06:14:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"db1d83c5aeb35c3a9e1f1fc2e782c5402d09f84f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 11:13:16.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a97276e26c97416b9396ea77c2e6fd9e : SUCCESS in 4m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51fe95bd2bfe4c0593dc785bff6cf61e : SUCCESS in 4m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57043a1787344ef8bbf18bde543168b8 : SUCCESS in 5m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/35fa2ea9327a43668b95ba28607111ce : SUCCESS in 11m 23s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe32c0f8e0fd4168bb7b08a19dd0ded8 : SUCCESS in 11m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdd987db0fb441caa103087eaf5c54f9 : SUCCESS in 13m 51s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb528d12bbc74793a23eb1b131bd44e5 : SUCCESS in 11m 33s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c186e1390ec4bf7bd64092347edac3e : SUCCESS in 1h 21m 22s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3af0b3caf5224701a4bceacebc300d5b : SUCCESS in 1h 26m 47s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/60bcb05772854939892f3a44f8333abf : SUCCESS in 1h 23m 54s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/224fd529627f425eae43effefa2289b8 : SUCCESS in 56m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3b12f603c124f89827ff6ec0bcc2ab1 : SUCCESS in 4m 56s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/65cae24e592b4be88f8f65b3a43b3ec8 : SUCCESS in 1h 52m 54s","accounts_in_message":[],"_revision_number":5},{"id":"3f354c7c24f5215fff5ed79312974fe134b3c954","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 11:13:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"eb5bd8e33020ca93606dfa9d188a346cd3def6dc","revisions":{"0675c2c58896dc6f8ffdc66f981f304ecc01004b":{"kind":"REWORK","_number":1,"created":"2021-08-29 02:32:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/806485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/85/806485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/1"}}},"commit":{"parents":[{"commit":"cd984ba49f746b603facbc7736b7e26f6f0a0c1d","subject":"Merge \"Support Puppet 7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/cd984ba49f746b603facbc7736b7e26f6f0a0c1d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 02:31:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 02:31:00.000000000","tz":540},"subject":"cache: Allow customizing status of the backend package","message":"cache: Allow customizing status of the backend package\n\nThis change introduces the package_ensure parameter to oslo::cache\nresource type, so that users can define status of the backend package.\n\nChange-Id: I34f10b51e1179beed327b9544b0899105cb30895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/0675c2c58896dc6f8ffdc66f981f304ecc01004b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/0675c2c58896dc6f8ffdc66f981f304ecc01004b"}]},"branch":"refs/heads/master"},"8173118bf88b908b93140f21f41e33bb5e039386":{"kind":"REWORK","_number":2,"created":"2021-08-29 05:31:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/806485/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/85/806485/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/2"}}},"commit":{"parents":[{"commit":"cd984ba49f746b603facbc7736b7e26f6f0a0c1d","subject":"Merge \"Support Puppet 7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/cd984ba49f746b603facbc7736b7e26f6f0a0c1d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 02:31:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 05:30:54.000000000","tz":540},"subject":"cache: Allow customizing status of the backend package","message":"cache: Allow customizing status of the backend package\n\nThis change introduces the package_ensure parameter to oslo::cache\nresource type, so that users can define status of the backend package.\n\nChange-Id: I34f10b51e1179beed327b9544b0899105cb30895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/8173118bf88b908b93140f21f41e33bb5e039386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/8173118bf88b908b93140f21f41e33bb5e039386"}]},"branch":"refs/heads/master"},"f1fb8f0695be43995df50897568c00df2ed9ac51":{"kind":"REWORK","_number":3,"created":"2021-08-29 05:34:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/806485/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/85/806485/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/3"}}},"commit":{"parents":[{"commit":"cd984ba49f746b603facbc7736b7e26f6f0a0c1d","subject":"Merge \"Support Puppet 7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/cd984ba49f746b603facbc7736b7e26f6f0a0c1d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 02:31:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 05:33:46.000000000","tz":540},"subject":"cache: Allow customizing status of the backend package","message":"cache: Allow customizing status of the backend package\n\nThis change introduces the package_ensure parameter to oslo::cache\nresource type, so that users can define status of the backend package.\n\nChange-Id: I34f10b51e1179beed327b9544b0899105cb30895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f1fb8f0695be43995df50897568c00df2ed9ac51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f1fb8f0695be43995df50897568c00df2ed9ac51"}]},"branch":"refs/heads/master"},"331635c04d24afe8dda0ffed78c58ae0ba547882":{"kind":"REWORK","_number":4,"created":"2021-08-29 06:06:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/806485/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/85/806485/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/4"}}},"commit":{"parents":[{"commit":"cd984ba49f746b603facbc7736b7e26f6f0a0c1d","subject":"Merge \"Support Puppet 7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/cd984ba49f746b603facbc7736b7e26f6f0a0c1d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 02:31:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 06:06:16.000000000","tz":540},"subject":"cache: Allow customizing status of the backend package","message":"cache: Allow customizing status of the backend package\n\nThis change introduces the package_ensure parameter to oslo::cache\nresource type, so that users can define status of the backend package.\n\nChange-Id: I34f10b51e1179beed327b9544b0899105cb30895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/331635c04d24afe8dda0ffed78c58ae0ba547882"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/331635c04d24afe8dda0ffed78c58ae0ba547882"}]},"branch":"refs/heads/master"},"eb5bd8e33020ca93606dfa9d188a346cd3def6dc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-30 08:01:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/806485/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/85/806485/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/85/806485/5"}}},"commit":{"parents":[{"commit":"b05244e2a1271ed71884d63421ec6746284d360c","subject":"Merge \"[oslo.middleware] Add support for basic auth middleware options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b05244e2a1271ed71884d63421ec6746284d360c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 02:31:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-30 08:01:13.000000000","tz":0},"subject":"cache: Allow customizing status of the backend package","message":"cache: Allow customizing status of the backend package\n\nThis change introduces the package_ensure parameter to oslo::cache\nresource type, so that users can define status of the backend package.\n\nChange-Id: I34f10b51e1179beed327b9544b0899105cb30895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/eb5bd8e33020ca93606dfa9d188a346cd3def6dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/eb5bd8e33020ca93606dfa9d188a346cd3def6dc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
