)]}'
{"id":"openstack%2Fpuppet-oslo~806486","triplet_id":"openstack%2Fpuppet-oslo~master~I459b7b752a83f75b1cb05c2104812133039a9162","project":"openstack/puppet-oslo","branch":"master","topic":"coordination-pkg","hashtags":[],"change_id":"I459b7b752a83f75b1cb05c2104812133039a9162","subject":"coordination: Allow disabling management of backend package","status":"MERGED","created":"2021-08-29 02:44:52.000000000","updated":"2021-09-20 11:15:03.000000000","submitted":"2021-09-20 11:13:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806486-coordination-pkg","meta_rev_id":"e5794ea5a89ad4ab020595d67053208573b6c67b","_number":806486,"virtual_id_number":806486,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-20 11:13:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-09-02 16:13:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-09-20 06:03:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-09-20 06:03:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-29 05:03:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-02 16:13:42.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-09-20 06:03:33.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"ab262b68079e54e036bf063a749c4495ab2afe2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-29 02:44:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4600f7f40c2fd552ff40f56f12ae5fcee4a8d014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 05:03:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/534753b3e4a3422ea3403f765f1c921a : SUCCESS in 4m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe7659f7a9b74bfdb1284562e38aed85 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e4e9f41f7094823a3b0cda500eb5032 : SUCCESS in 4m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/678a6351af4c475f81d99ab12fccdca8 : SUCCESS in 11m 48s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d6a90e58c7a490c8e8607b959aa6a12 : SUCCESS in 10m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e32619d156d3464ea43eda3bd8311b57 : SUCCESS in 8m 19s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1c1e6e938684eef8ac1e614162e51bf : SUCCESS in 8m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad1a382833ae4bee9ebc2975282ad60c : SUCCESS in 8m 31s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3efc29b961a446ddab82105b45f649a9 : SUCCESS in 50m 07s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/113f1c1a2cbf4a61af099fd49ddfa401 : SUCCESS in 51m 19s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4466a51d7e7d4a61852b96c8e968f226 : FAILURE in 30m 23s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d295275aed0944c2810a988b8a967072 : SUCCESS in 39m 07s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3044041381b404789c74795744d684d : FAILURE in 15m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba83a82371144a47ad7c4d3399108b48 : SUCCESS in 1h 08m 56s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8c0b51071f5436d9c551c81da5cd19b : SUCCESS in 1h 17m 21s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd1ef84fb45847cfa331a70fda24a7b7 : SUCCESS in 1h 11m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ec72503a5bd43d39c5122aee740f13d : SUCCESS in 1h 08m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09a5486911de4c9d944a86d001aafb4f : SUCCESS in 4m 23s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e58a77bc6851443589474b2939673486 : SUCCESS in 2h 15m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f33f60159c2bd06595e50b810e202516634c943f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-29 05:27:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"67d67a0d21163a399056edf02196f51f9979071c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-29 05:35:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bfc18712490acca89cd86aec5e7f89b6959bce69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 07:37:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c65cb7a59854390a2e2c646f6778442 : SUCCESS in 4m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c20226da889482888f5327fd7dd74ba : SUCCESS in 3m 56s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a2af283f01540588c751b8ab0cc40c3 : SUCCESS in 3m 35s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6eb807e13000476d9f4374c6ff20a798 : SUCCESS in 10m 38s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ffecf02833449bb8abc3ccfabf8117e : SUCCESS in 10m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/685d9dccfa5848e9bee7ca3d9062ef58 : SUCCESS in 8m 15s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/261452a58e96407495054ff697c55b66 : SUCCESS in 7m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbbc68e0035f4dcfa5dd9df97efb18e6 : SUCCESS in 7m 49s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/834f1a9bb3284cc19df1052505502f8f : SUCCESS in 51m 12s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d08f42d83b604931910240d9f67eaca3 : SUCCESS in 53m 27s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3910b8812a90472eb7c6df2a6b69276f : FAILURE in 34m 05s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/139383e6c24547b881e7919b020ee785 : SUCCESS in 34m 10s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a54e2f3c9204f6fa8f0bc6d5bf8c80a : SUCCESS in 21m 18s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a3eb905b7b64612b36d79cda0291c9f : SUCCESS in 1h 19m 44s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ab74b71624e4d069dabc499d47384c0 : FAILURE in 55m 39s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f64a2f0d7af04e5ca5b81ab2454df31d : SUCCESS in 1h 28m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/42512cbdb0d74d159813b0acd11f1909 : SUCCESS in 52m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d8389253e124d29841d4516446aebb3 : SUCCESS in 4m 05s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c1e8ce2fcffd4c958c3ae99e84a0f7c1 : FAILURE in 1h 55m 15s","accounts_in_message":[],"_revision_number":3},{"id":"7f516b87e22ab7e638fc3eea4d82295d5b3c9192","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-29 08:36:15.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"06913bf9228b0687eea742bd2a5f16e703e6185e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 10:10:30.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ff34ce5a46645ea800465be7bb98d1f : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb6d0beb416440849b82f11de81ba840 : SUCCESS in 4m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd12b6cdb17e48c0bfa01558c18b99c5 : SUCCESS in 3m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ff45d22b0ee4ef0b89d60198561cc00 : SUCCESS in 12m 02s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3b5e19c3ac94378838efb2eb6518609 : SUCCESS in 9m 13s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5a59662c54c498989caadda1da1b8ca : SUCCESS in 7m 59s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/586d2ac1b51d4f93aa49e5b7389f2891 : SUCCESS in 8m 29s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/625a6ab4671243b78f684ff4ebd4c92b : SUCCESS in 9m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05360dc35fb241d99f941bef79e5508d : SUCCESS in 45m 02s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6179be9165824efea050b3cd70b38375 : SUCCESS in 53m 00s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b8b9f698e774505a4767fe34da3f795 : FAILURE in 27m 30s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb9dec0fa02f4a9ea82119bb132ffc81 : SUCCESS in 40m 27s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a317c0312f0c417c917e91c4604b46f3 : SUCCESS in 23m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/24ae426769d04ae8a584a290ecc228ae : SUCCESS in 1h 18m 04s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/053c9c30f34c4be5bac26f03e0a89b84 : SUCCESS in 1h 10m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/71ffb58a7ff247f29f705edcbec9925c : SUCCESS in 1h 27m 33s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5900aa5004b14c2888a1761cb58ce96c : SUCCESS in 57m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85a6375f57014a59ba3f11f8353df333 : SUCCESS in 4m 05s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/69d61a362a284745b1fd6fc484c2ff27 : FAILURE in 1h 20m 45s","accounts_in_message":[],"_revision_number":3},{"id":"6398418ced0f9a1e8d9292778a912052d33ca4ab","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-30 10:12:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"58600217394fcdf77681d5d1e9acf6adc69e090b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 12:12:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1c4120c2c8442f38b0d4e8854798230 : SUCCESS in 5m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec7c766fcb724e288a0e2d6dbe5a357d : SUCCESS in 4m 15s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/095d258e9a344b0e909b5d4e7d0e48a6 : SUCCESS in 5m 06s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/081793b7e75d4c82850008ca004f8407 : SUCCESS in 12m 15s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/27e1d273234a40a58fd53ce6414e939b : SUCCESS in 12m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ebd99c9d84b46dd86b9b9640ad79c4a : SUCCESS in 8m 58s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5792cba54bea4ee7adbafbdf466a5fed : SUCCESS in 9m 35s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77163a2e4316424a91d61b239cbfece5 : SUCCESS in 10m 33s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a3dbc4f74084c9d936702f08a993b28 : SUCCESS in 49m 33s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4707169881b94fb0be0c2a2e0125dce0 : SUCCESS in 48m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/786b156295704f14bed84fbfa4c821be : FAILURE in 28m 18s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2da26431551247ea999408a2326cff90 : SUCCESS in 39m 27s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d7a481828a844e6837018be3e4a17ec : SUCCESS in 23m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e190abaca9014314ac904a6777c566b7 : SUCCESS in 1h 17m 13s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d974fbcdd464426b9a182fd6ff29691 : SUCCESS in 1h 39m 24s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0262b1223d04c22b73e8ed233caf120 : SUCCESS in 1h 04m 35s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/56d46ebf78064d47a84a51532f2e253f : SUCCESS in 1h 00m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/852ea1519625438780b8f9f37006af62 : SUCCESS in 4m 29s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5dd6f76ce14a42e8afdb34cca749dca8 : SUCCESS in 1h 54m 42s","accounts_in_message":[],"_revision_number":3},{"id":"aa027f8821f85eb7778f25018fb26e78e17bcd91","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-02 16:13:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"496b30ec42f78c5eeff197ec630451f695eaef7f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-09-20 06:03:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"11edef57ece570d7c2bd3cb30918e2b194c45464","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 06:14:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"21840626013330f21294e17430a7927bb42ebe1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 11:13:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d90aa1797ca4172857a830561341134 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/104c7c7fee224a4f92c02f293ef641d2 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee32d84333414f4ca75a8d86a257d0d8 : SUCCESS in 5m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4bb95ed0a7fc489ba6dbafd2aa947531 : SUCCESS in 10m 58s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/41822f71146144e1b9343118d0697d26 : SUCCESS in 11m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd7f608951b542279d6b1cb70bd792ce : SUCCESS in 9m 18s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a159f8a32fb41e797067f4f6966c394 : SUCCESS in 11m 01s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3aaa2769545947aebefaefe71581ce8a : SUCCESS in 1h 50m 48s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4bcc3db0d284b1cb1ab0212b1815bf4 : SUCCESS in 1h 38m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/17ae2328be254dc2b6f691fcae1d9640 : SUCCESS in 1h 22m 53s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/975cfeab59a04c4395db767da9a123b4 : SUCCESS in 59m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6ee62a22a564294bafdfde996e45ef6 : SUCCESS in 4m 32s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0990d894f30a48ddaacf045bcca823e5 : SUCCESS in 2h 21m 08s","accounts_in_message":[],"_revision_number":3},{"id":"abd26510284ed3a0e1d02b868183eea42728e29c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 11:13:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e5794ea5a89ad4ab020595d67053208573b6c67b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 11:15:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51559deac01d423689bf003e4ba2ba46 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"872883d7f507a2d746c40ca92ee861ed27c96c62","revisions":{"b66d8ea26d526d577e7ef201c5949b151627c5eb":{"kind":"REWORK","_number":1,"created":"2021-08-29 02:44:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/806486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/86/806486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/806486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/806486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/806486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/86/806486/1"}}},"commit":{"parents":[{"commit":"b05244e2a1271ed71884d63421ec6746284d360c","subject":"Merge \"[oslo.middleware] Add support for basic auth middleware options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b05244e2a1271ed71884d63421ec6746284d360c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 02:42:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 02:44:39.000000000","tz":540},"subject":"coordination: Allow disabling management of backend package","message":"coordination: Allow disabling management of backend package\n\n... which is implemented in other resources like oslo::cache.\n\nChange-Id: I459b7b752a83f75b1cb05c2104812133039a9162\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b66d8ea26d526d577e7ef201c5949b151627c5eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b66d8ea26d526d577e7ef201c5949b151627c5eb"}]},"branch":"refs/heads/master"},"e4e0d3e391761a7e81392a5e491ab56e4b86c4c4":{"kind":"REWORK","_number":2,"created":"2021-08-29 05:27:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/806486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/86/806486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/806486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/806486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/806486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/86/806486/2"}}},"commit":{"parents":[{"commit":"b05244e2a1271ed71884d63421ec6746284d360c","subject":"Merge \"[oslo.middleware] Add support for basic auth middleware options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b05244e2a1271ed71884d63421ec6746284d360c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 02:42:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 05:27:31.000000000","tz":540},"subject":"coordination: Make status of the backend package configurable","message":"coordination: Make status of the backend package configurable\n\n... as is implemented in other resources like oslo::cache.\n\nChange-Id: I459b7b752a83f75b1cb05c2104812133039a9162\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/e4e0d3e391761a7e81392a5e491ab56e4b86c4c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/e4e0d3e391761a7e81392a5e491ab56e4b86c4c4"}]},"branch":"refs/heads/master"},"872883d7f507a2d746c40ca92ee861ed27c96c62":{"kind":"REWORK","_number":3,"created":"2021-08-29 05:35:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/806486/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/86/806486/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/806486/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/806486/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/806486/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/86/806486/3"}}},"commit":{"parents":[{"commit":"b05244e2a1271ed71884d63421ec6746284d360c","subject":"Merge \"[oslo.middleware] Add support for basic auth middleware options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b05244e2a1271ed71884d63421ec6746284d360c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 02:42:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 05:35:01.000000000","tz":540},"subject":"coordination: Allow disabling management of backend package","message":"coordination: Allow disabling management of backend package\n\n... which is already implemented in other resources like oslo::cache.\n\nChange-Id: I459b7b752a83f75b1cb05c2104812133039a9162\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/872883d7f507a2d746c40ca92ee861ed27c96c62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/872883d7f507a2d746c40ca92ee861ed27c96c62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
