)]}'
{"id":"openstack%2Fpuppet-oslo~807851","triplet_id":"openstack%2Fpuppet-oslo~master~I7deed1bca1ce2f715967e966d4b0cedc1f227027","project":"openstack/puppet-oslo","branch":"master","topic":"socket-keepalive","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-27 17:39:25.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2022-01-27 17:39:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7deed1bca1ce2f715967e966d4b0cedc1f227027","subject":"Add socket keepalive options for the pymemcache backend","status":"MERGED","created":"2021-09-08 10:20:09.000000000","updated":"2022-01-27 17:42:40.000000000","submitted":"2022-01-27 17:39:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807851-socket-keepalive","meta_rev_id":"97c728a938a58850c3b31f033cdd1d72b3fa9541","_number":807851,"virtual_id_number":807851,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-27 17:39:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-01-27 11:08:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-01-27 15:06:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-01-27 15:06:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-08 10:20:36.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-08 11:04:29.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-09-08 12:04:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-05 09:54:11.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2021-11-10 16:31:06.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-11-12 09:53:38.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"230092945873c95ae27897242ccfc7968744ce8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-09-08 10:20:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39bc250e972b1790c5d974b2a0c0f2b54079564d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-08 11:04:29.000000000","message":"Patch Set 1: Code-Review-1\n\nIt seems the dependent change in oslo.cache has not yet merged, so let me put -1 atm to avoid accidental merge.","accounts_in_message":[],"_revision_number":1},{"id":"dbc7faca120ad62e58889d18d0021c66469fa9aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 12:04:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c303f21a55584e1ea31e26ce7adeb6a9 : FAILURE in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ba7dd968d594e81a9db363de2d4bbe6 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d31bedbeeea4cb29fec71f6305eb168 : SUCCESS in 4m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d9968c0bcd945fa95492fb877302fc1 : FAILURE in 13m 06s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ddd32b7c4f544f3ebc08b6c5ac2c9666 : FAILURE in 14m 13s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42e00a8b3993424ead7a2ac4b3a0ccbb : FAILURE in 11m 01s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/656312497c384cba9d5e87d1da1561c5 : FAILURE in 10m 54s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ff2f0a712324b98892a6f52945123e5 : FAILURE in 10m 33s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d826f3bbff7848e59b48079e3258cac7 : SUCCESS in 57m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a01154ca9e1e4eeca102a33783cf3993 : SUCCESS in 57m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8923c388c1d343d6afb79be7721ed225 : FAILURE in 27m 38s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db150d9745a1419eaf61e9d3b82320ce : SUCCESS in 46m 34s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/855f8be68bf04e5b8ad3b4e8ee590b16 : SUCCESS in 20m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/78919f21431747dbb2e65c597591d9db : SUCCESS in 1h 14m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/694f825867c44e3889995788a07e8d7b : SUCCESS in 1h 18m 53s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f25cfb7ef4964a56a17dece61c01f390 : SUCCESS in 1h 21m 05s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8532ccef7a1845b7829212b77a9d5ce4 : SUCCESS in 1h 02m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/137dec8b612a405383fea641ca17785c : SUCCESS in 4m 44s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1ae5341eb4ea4c859668711a9d906a9f : SUCCESS in 1h 42m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2ae1247275c7347727c385fb961a1c59403083e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-09-14 11:53:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d4cd000a2e45f587f1087fce21075c1fad0d737c","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-09-14 11:54:15.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e It seems the dependent change in oslo.cache has not yet merged, so let me put -1 atm to avoid accidental merge.\n\nI put a Depends-On on the relevant patch, this should let us keep track of things.","accounts_in_message":[],"_revision_number":1},{"id":"a8f9aa13b4bb1f380b9603faa65ab643913de747","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-09-14 11:56:38.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"8ac1b7ec97c50dddfb10b2836a08d1bbcc056a50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 14:01:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22dc97f63a7b4e2da0e5cdd12e6a94de : FAILURE in 5m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/277336dfaffc44b6ad2d9984a4780803 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/384ed32eb35248c9968dc70d6dcad422 : SUCCESS in 4m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/787248e27144492f99f0d8e2c6d074ca : FAILURE in 9m 59s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0329a550b92141d89a9f3e20dd6fbe14 : FAILURE in 10m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ea8c388bbf64ab7bd6ccbc8bc43e831 : FAILURE in 10m 50s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7391076347cc490496af6495fa23ecde : FAILURE in 7m 35s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9e886ad94604b1fbad51a62d4c5b60d : FAILURE in 8m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33bdefe0c58d43f98f54b1b1d7ffd970 : SUCCESS in 53m 18s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15cdb1844b1143f0981dd24c2239a7cb : SUCCESS in 47m 54s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bac85fd5031249179410a9e8e4144379 : FAILURE in 34m 30s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddc22f40fd5f4345878e040b3853283e : SUCCESS in 57m 46s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/61c9c6fef8f744d89bcdd90dd2453a4f : SUCCESS in 22m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2bcbfa3a8524bfdaa1e70769cdd3e97 : SUCCESS in 1h 14m 16s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/75342de34e8947c4b858226647617cab : SUCCESS in 1h 58m 30s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/88ce409f59ba47ea85af76c4e6fe6895 : SUCCESS in 1h 17m 27s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d50debba4e5d4d9d9dc417aa4c6df576 : SUCCESS in 1h 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96bf3994ed0c48f3a38328307b52c465 : SUCCESS in 4m 05s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0ac8819387df45e192700293ce1d67a6 : SUCCESS in 1h 53m 48s","accounts_in_message":[],"_revision_number":3},{"id":"bc98b40453dc6dbb206d8258f37a2c51f129974a","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-11-03 08:36:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"41bc38035c94ca03531eb49b2a5f8c8d3484fd40","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-11-03 08:37:12.000000000","message":"Topic set to socket-keepalive","accounts_in_message":[],"_revision_number":3},{"id":"861128289986c596b433de9ae0bfdfeda623ea0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 10:35:51.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dea8099d68db4f50a862dd78638610de : FAILURE in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94cb3672ea714d42847e95614bd07c86 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe911bf90c60423eab883a9073de6621 : SUCCESS in 4m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/88bbc968d4e84cfd9bb59d7ba8e56d55 : FAILURE in 13m 15s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3711c7927b94638b31c1162b11aaacf : FAILURE in 10m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b9004b300024227856e3212afb2cea1 : FAILURE in 7m 18s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a7fe62fd7054a129c92f92bab09ad8f : FAILURE in 7m 36s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95690fa8f8254476966549a26cb30c21 : FAILURE in 7m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/161fa555ed124e28a426e4644e3a6e33 : FAILURE in 38m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d2bc5065a8c43ce9863d814eb1a7228 : FAILURE in 33m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ea326170f56458b9d03b1315721a415 : FAILURE in 1h 07m 31s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f3f5225a27f45f1bb98566be3699b68 : FAILURE in 34m 36s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f1cffa4ac654cf4a048b88598a743cf : SUCCESS in 25m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c731501508b44f87a05ebe28567e5dd5 : SUCCESS in 1h 29m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c4786428a85423582a30f1a7916f66e : SUCCESS in 1h 44m 17s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b91d028f373640d3a253545032749231 : SUCCESS in 1h 17m 43s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9bd74c66b1647b38917d71b686c031e : SUCCESS in 1h 15m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b081197031174896b44cf2607f96f48e : SUCCESS in 4m 00s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/07048e522d04485283bc6540f731b6b8 : SUCCESS in 1h 57m 04s","accounts_in_message":[],"_revision_number":3},{"id":"95d485fd4b8db287a42cfe878040a2723a3ffef5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-03 13:27:04.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f685d24b9c816bee8d2adfaad421de927dcf37d9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-03 13:27:16.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a78c17e49e1240e0800fbd0a27a85f9e9e7e30e4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-03 13:28:07.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cf6dae03d4ce39e42ea18474cba01e5f18de42b0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-03 13:29:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ad2614094702d23abf58c1eaa07cc23ae961d872","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-11-03 13:35:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"045852c76477695b214147d75810c397e497c673","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-11-03 13:36:22.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d664f52e7695f6f98f1aedd819b0c610f565b896","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 15:39:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a12dcc48b7f640b5be52e84128fbe837 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e291424e966432cb02fc62e64ed3d01 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e42e93c1c8b4526a4275ed3a688d168 : SUCCESS in 4m 07s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d6912bf2c6694de380be78ba9a8bcf3a : SUCCESS in 12m 10s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/deae38c88f174c6d9f799a6146891b96 : SUCCESS in 11m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dcbb0ff0abd409ba977fa06c02b4049 : SUCCESS in 8m 54s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1f33ccaf8f14d4cb0534030ef786370 : SUCCESS in 8m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edd427435a58412ba298a259ed9463fa : SUCCESS in 8m 46s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06a990f2547a47038208dc2e2527fe74 : FAILURE in 43m 48s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27adc769de8f43599eff5bc0201178f7 : FAILURE in 44m 19s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f41181855a843febd15dbdada4576f4 : FAILURE in 41m 39s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b4163e95b924c41b7bf4b176d2cc131 : FAILURE in 27m 10s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/160e33aa42574843a517c24b3e5be730 : SUCCESS in 27m 42s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0089d5378970473b97a6c3792fdf816a : SUCCESS in 1h 15m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c10cac0760d455e881565ac8c9dea8a : SUCCESS in 1h 34m 33s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c4052030455485f86d9319aae5fa9ca : SUCCESS in 1h 24m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/58811af29e0e43fcbf2e92904e87ef5f : SUCCESS in 1h 06m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e79d1d8cb6d4cd5b61664855775919b : SUCCESS in 4m 30s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5934898aa3484030a6e236aec84ad7b1 : SUCCESS in 2h 00m 31s","accounts_in_message":[],"_revision_number":4},{"id":"a7fbf4987919ae494e2fb2d772f148a32fa630ee","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-11-05 09:54:11.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"7ed0758728c99622346ddcf656047f954328e301","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-10 16:31:06.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"fa70679f17f2381da4c1aba7ba5fa1abd8f0d80c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-12 09:53:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"61bf1423b4b87d8ff8f7a570b50b445af228d798","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-23 16:46:41.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"593ccda3fd90e06ccf8e1a1510e7c2223e747f52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-12-21 09:48:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a8e34b40af93a9edbf64fb3077e08982aac438d8","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-12-21 10:55:52.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"50d861d00527bfe3cc39d01ac9eb1bf8a887a613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 11:51:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6a772b8335e49ce9d38c5acb68c314b : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2763952fafe9409c9b61765b67fbe856 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/962f9e1e511540ce8af520ac5fdd90b6 : SUCCESS in 3m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d649be018dc4a31a6a40ca1d28c52ea : SUCCESS in 11m 47s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ebc8293044f404097006dbf8b39820e : SUCCESS in 11m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84abff49fed142bb9ae5f10be5771f6d : SUCCESS in 8m 23s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60a9404ef2c04d28aa242472a9ce7769 : SUCCESS in 8m 47s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7075adeb394d4b348852c518caa0e80b : SUCCESS in 8m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7df16c1b2eb444cd964658b045342913 : FAILURE in 32m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e15ee263eb104df0ae7aa9aef2c43706 : FAILURE in 39m 21s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c002f51e97b94c3583068cef6a4d46b7 : FAILURE in 54m 43s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4abd23be7e3492ab47c5b7225fda747 : FAILURE in 23m 31s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f7c3a78f548455db1006a0c357f1aee : SUCCESS in 32m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c28e144ad3b74a6aa05b6208b168b044 : SUCCESS in 1h 12m 52s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a66c924fbbe4fe49163669819d82ac8 : SUCCESS in 1h 35m 29s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b6441f2fce4b4ddc9890888369802c98 : SUCCESS in 1h 15m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a2f5717f56a4d6a842a86fa1c3444f9 : SUCCESS in 57m 15s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d40ed5ab4a948f9b319fcbb9ac36691 : SUCCESS in 39m 52s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3f4f2c722564815b5303b7571d69858 : SUCCESS in 47m 07s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc17f5b62026423995f78704e8aa867b : SUCCESS in 52m 45s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/add9f631a6e74f3a81604e45a4eb2e6a : SUCCESS in 56m 02s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/74e52273cf1e42f8a1bec40274ac228e : SUCCESS in 40m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8ea2a75f0c24d5da1eba7096571910a : SUCCESS in 3m 53s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/901b8336daf24dc4ac1ad7efae17b61d : SUCCESS in 1h 58m 53s","accounts_in_message":[],"_revision_number":5},{"id":"deb5a6286da4cb6e7c88a18470cd2380c442f80b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-27 04:19:39.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7037d1a3c9ce696519c9b01bb576316ae152e6ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-27 04:21:21.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e62c0910f1519b2d4f1d7efe8a7f937bb8b83b13","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-27 04:21:56.000000000","message":"Patch Set 6: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b3d07a16673d135bfaf78f225293a605924c877e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 06:32:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cf4c5247cda4f18bd99889f37be91fa : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8f38d76ab4a459398fcbea9a0ba5f2e : SUCCESS in 4m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66d63218d3f542b2a7fb40cfd5e4cc90 : SUCCESS in 3m 59s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3c7198e48534be9a2e129fc304e6425 : SUCCESS in 13m 46s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4c78604da9e4c728ada455172ec9612 : SUCCESS in 9m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a380b504f24d410a999ba582c4e02e04 : SUCCESS in 10m 20s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e677aa6db594773b6ee4d42955378bc : SUCCESS in 7m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc319c43f430407ba664540c2d916a43 : SUCCESS in 11m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5e6de6503b2490781bf03614e721eff : FAILURE in 33m 42s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26f6c5a0edce45f3b67a246e9e8df5a7 : FAILURE in 40m 56s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ecd35c25c844eb593e8c1318427183a : FAILURE in 22m 45s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67ba3076ae6d40859fb2d01d7ab2b644 : FAILURE in 27m 12s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/84ec40bef950445b9f0da70474fc6a31 : SUCCESS in 23m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c61ed58090f041eea5888633c9a8ff3d : SUCCESS in 1h 11m 59s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b79fa5a7cd34fccb8639babb7b272bd : SUCCESS in 1h 23m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a72b9aa25f6f44cd99f53b8ead29d9db : SUCCESS in 1h 21m 42s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b5f93b059fd4ca58207ecb09ff03c72 : SUCCESS in 59m 09s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/93651a591182491ab43ef177eedc2b68 : RETRY_LIMIT in 7m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/577aa62435a2410eb424b0e881a7cd0b : RETRY_LIMIT in 8m 07s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f6402f0e1bf4ab7b47da743c25329df : RETRY_LIMIT in 5m 10s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a0bbd7d29824930ab2c36f14a9eba98 : RETRY_LIMIT in 5m 34s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/691884c7b18040c486b4d5c5dfe5ce33 : RETRY_LIMIT in 5m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84ae6a3579e14f53b73f971ff480a419 : SUCCESS in 3m 52s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/675842870dcc4c7494d4ff74d407af2a : SUCCESS in 2h 08m 29s","accounts_in_message":[],"_revision_number":6},{"id":"c46eb952d8f5b61c8dec5facc7b2732e597b855b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-27 11:07:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"88cffa597af9107d2cf8b2e952da144104ece262","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-27 11:08:04.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bc1f0aec545547b94321619e7f78a278c0f39d6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 13:11:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74159909f73e4d208cc9462ebc194182 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62532a96d20548c18ace1c71f48ed50e : SUCCESS in 3m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99363e8a1e984002bb480209a92fd0a7 : SUCCESS in 4m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a10a6df95014803bd7eb71ef66fe0a4 : SUCCESS in 11m 01s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7cd29cbe5ba4d01a5381467f8d6a84d : SUCCESS in 10m 57s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24402f282b924ec8be55078442d8daa8 : SUCCESS in 8m 25s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a25c4bf12f4417385e3034b682d2fa5 : SUCCESS in 8m 32s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86e1c3c961f9461cb9ac1719c89ca910 : SUCCESS in 8m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f49a56b3867449ed9ba4aad89314b88d : FAILURE in 37m 32s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18432796360e4010abee9e57ae56dfd8 : FAILURE in 39m 13s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f7341bd57314a1e84400d13f6c0f385 : FAILURE in 29m 55s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1987d1bf2bd54eaf81fd92b829067d74 : FAILURE in 33m 10s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d5d2a892c0434ffe8bbd3e6b18c9d543 : SUCCESS in 25m 03s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d774ee535eef4a9b97e797f646bcd24a : SUCCESS in 1h 20m 44s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0c2fe27566c4ff387a577e3283c9f65 : SUCCESS in 1h 32m 04s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ccfdee274954040a5f0b57a1abfe0bf : SUCCESS in 1h 23m 19s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6cc1d034b18140ac93256f1c584a780e : SUCCESS in 1h 02m 10s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/80309ea25d3241e9aeb61c3c0e2c44e4 : SUCCESS in 25m 03s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fdd569ac5fca4dee8f5c99b90ba44eeb : FAILURE in 53m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ac75d7eb95a48b6b7fb4caa4504b15a : FAILURE in 51m 11s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/56894187f193478d98d34141172cc9ca : FAILURE in 1h 02m 36s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f78a52e906664c53b7d639051e75f8e4 : FAILURE in 50m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aac9d2d872754fcc8f952995fbb61094 : SUCCESS in 4m 09s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/12ce5abdb4234d6f85378480d934786a : SUCCESS in 2h 00m 41s","accounts_in_message":[],"_revision_number":7},{"id":"bd6cc0ab08acea5399ada537388f74c31db10a2b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-27 15:06:19.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"0f4fc7db910cf9a9062fab27d1f44e946cc4da1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 15:07:47.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"c6fa3c29e90fbc40e07f24cb31615d37d8072c45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 17:39:25.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/289f719ea8eb47fb921bbe345083e761 : SUCCESS in 5m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d7f60ff476e485e8077809dd93174a2 : SUCCESS in 5m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/076941e937a74bc2b80997f7e88b4794 : SUCCESS in 29m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/15a3274d16cf40358443967309e2647b : SUCCESS in 13m 25s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/08f020c9cdc840b1b3f5279177121048 : SUCCESS in 10m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e77de2c8bb54fb19c01839be86f695f : SUCCESS in 10m 29s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d148abf9eab14fa2a8a1c9be1fe31b3a : SUCCESS in 8m 45s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2cadcca4b1d34ea58bd4dd91a35a69d7 : SUCCESS in 1h 26m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/945c2d80674142198e4378388c8abdfb : SUCCESS in 1h 40m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1447ede0f8a440608d029115d9005ec3 : SUCCESS in 1h 31m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dcd8067d2ec54301b6926228f79cfb5e : SUCCESS in 1h 13m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f20be6ae01e2414db219d439978b16ad : SUCCESS in 5m 04s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0957aa2f33804b288c9a01cc6b8c429d : SUCCESS in 2h 25m 54s","accounts_in_message":[],"_revision_number":7},{"id":"3b299999a48a4cb4de2d2d0a65a38ce96dc26531","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 17:39:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"97c728a938a58850c3b31f033cdd1d72b3fa9541","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 17:42:40.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/616ac24ad1a544b282e7b07fd76381bd : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"60495a42c9e46144dc3625cef6ede24e18245011","revisions":{"77787ddba15bb0f0b6079c9483beb4a5af3cc468":{"kind":"REWORK","_number":1,"created":"2021-09-08 10:20:09.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/51/807851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/51/807851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/1"}}},"commit":{"parents":[{"commit":"b05244e2a1271ed71884d63421ec6746284d360c","subject":"Merge \"[oslo.middleware] Add support for basic auth middleware options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b05244e2a1271ed71884d63421ec6746284d360c"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-09-08 10:16:11.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-09-08 10:16:11.000000000","tz":120},"subject":"Add socket keepalive options for the pymemcache backend","message":"Add socket keepalive options for the pymemcache backend\n\nThis patch specifies a set of options required to setup the socket\nkeepalive feature of pymemcache (dogpile.cache) cache backend.\n\nOriginal oslo.cache change:\nhttps://review.opendev.org/c/openstack/oslo.cache/+/803716\n\nChange-Id: I7deed1bca1ce2f715967e966d4b0cedc1f227027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/77787ddba15bb0f0b6079c9483beb4a5af3cc468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/77787ddba15bb0f0b6079c9483beb4a5af3cc468"}]},"branch":"refs/heads/master"},"ab6abaf781bfe3acf59c3bacdc7550d88c08f2fc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-09-14 11:53:22.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/51/807851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/51/807851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/2"}}},"commit":{"parents":[{"commit":"b05244e2a1271ed71884d63421ec6746284d360c","subject":"Merge \"[oslo.middleware] Add support for basic auth middleware options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b05244e2a1271ed71884d63421ec6746284d360c"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-09-08 10:16:11.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-09-14 11:52:33.000000000","tz":120},"subject":"Add socket keepalive options for the pymemcache backend","message":"Add socket keepalive options for the pymemcache backend\n\nThis patch specifies a set of options required to setup the socket\nkeepalive feature of pymemcache (dogpile.cache) cache backend.\n\nOriginal oslo.cache change:\nhttps://review.opendev.org/c/openstack/oslo.cache/+/803716\n\nDepends-On: https://review.opendev.org/807851\nChange-Id: I7deed1bca1ce2f715967e966d4b0cedc1f227027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/ab6abaf781bfe3acf59c3bacdc7550d88c08f2fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/ab6abaf781bfe3acf59c3bacdc7550d88c08f2fc"}]},"branch":"refs/heads/master"},"c74772580c72abbad14672a356254318782b7515":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-09-14 11:56:38.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/51/807851/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/51/807851/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/3"}}},"commit":{"parents":[{"commit":"b05244e2a1271ed71884d63421ec6746284d360c","subject":"Merge \"[oslo.middleware] Add support for basic auth middleware options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b05244e2a1271ed71884d63421ec6746284d360c"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-09-08 10:16:11.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-09-14 11:56:22.000000000","tz":120},"subject":"Add socket keepalive options for the pymemcache backend","message":"Add socket keepalive options for the pymemcache backend\n\nThis patch specifies a set of options required to setup the socket\nkeepalive feature of pymemcache (dogpile.cache) cache backend.\n\nOriginal oslo.cache change:\nhttps://review.opendev.org/c/openstack/oslo.cache/+/803716\n\nDepends-On: https://review.opendev.org/803749\nChange-Id: I7deed1bca1ce2f715967e966d4b0cedc1f227027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c74772580c72abbad14672a356254318782b7515"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c74772580c72abbad14672a356254318782b7515"}]},"branch":"refs/heads/master"},"73430e436024cf949002ee15bafe1e533c966e60":{"kind":"REWORK","_number":4,"created":"2021-11-03 13:35:48.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/51/807851/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/51/807851/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/4"}}},"commit":{"parents":[{"commit":"b05244e2a1271ed71884d63421ec6746284d360c","subject":"Merge \"[oslo.middleware] Add support for basic auth middleware options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b05244e2a1271ed71884d63421ec6746284d360c"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-09-08 10:16:11.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-11-03 13:35:39.000000000","tz":60},"subject":"Add socket keepalive options for the pymemcache backend","message":"Add socket keepalive options for the pymemcache backend\n\nThis patch specifies a set of options required to setup the socket\nkeepalive feature of pymemcache (dogpile.cache) cache backend.\n\nOriginal oslo.cache change:\nhttps://review.opendev.org/c/openstack/oslo.cache/+/803716\n\nDepends-On: https://review.opendev.org/803749\nChange-Id: I7deed1bca1ce2f715967e966d4b0cedc1f227027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/73430e436024cf949002ee15bafe1e533c966e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/73430e436024cf949002ee15bafe1e533c966e60"}]},"branch":"refs/heads/master"},"19c00ea890d22e8e48cc983645c3469007eeac9b":{"kind":"REWORK","_number":5,"created":"2021-12-21 09:48:14.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/51/807851/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/51/807851/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/5"}}},"commit":{"parents":[{"commit":"b05244e2a1271ed71884d63421ec6746284d360c","subject":"Merge \"[oslo.middleware] Add support for basic auth middleware options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b05244e2a1271ed71884d63421ec6746284d360c"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-09-08 10:16:11.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-12-21 09:45:21.000000000","tz":60},"subject":"Add socket keepalive options for the pymemcache backend","message":"Add socket keepalive options for the pymemcache backend\n\nThis patch specifies a set of options required to setup the socket\nkeepalive feature of pymemcache (dogpile.cache) cache backend.\n\nDepends-On: https://review.opendev.org/803716\nChange-Id: I7deed1bca1ce2f715967e966d4b0cedc1f227027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/19c00ea890d22e8e48cc983645c3469007eeac9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/19c00ea890d22e8e48cc983645c3469007eeac9b"}]},"branch":"refs/heads/master"},"5f79034cdba5bd1d5e6327a5afb7d9adb41785ea":{"kind":"REWORK","_number":6,"created":"2022-01-27 04:21:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/807851/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/51/807851/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/6"}}},"commit":{"parents":[{"commit":"b05244e2a1271ed71884d63421ec6746284d360c","subject":"Merge \"[oslo.middleware] Add support for basic auth middleware options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b05244e2a1271ed71884d63421ec6746284d360c"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-09-08 10:16:11.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-27 04:21:12.000000000","tz":540},"subject":"Add socket keepalive options for the pymemcache backend","message":"Add socket keepalive options for the pymemcache backend\n\nThis patch specifies a set of options required to setup the socket\nkeepalive feature of pymemcache (dogpile.cache) cache backend.\n\nDepends-On: https://review.opendev.org/803716\nChange-Id: I7deed1bca1ce2f715967e966d4b0cedc1f227027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/5f79034cdba5bd1d5e6327a5afb7d9adb41785ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/5f79034cdba5bd1d5e6327a5afb7d9adb41785ea"}]},"branch":"refs/heads/master"},"60495a42c9e46144dc3625cef6ede24e18245011":{"kind":"REWORK","_number":7,"created":"2022-01-27 11:07:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/807851/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/51/807851/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/51/807851/7"}}},"commit":{"parents":[{"commit":"b05244e2a1271ed71884d63421ec6746284d360c","subject":"Merge \"[oslo.middleware] Add support for basic auth middleware options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b05244e2a1271ed71884d63421ec6746284d360c"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-09-08 10:16:11.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-27 11:07:32.000000000","tz":540},"subject":"Add socket keepalive options for the pymemcache backend","message":"Add socket keepalive options for the pymemcache backend\n\nThis patch specifies a set of options required to setup the socket\nkeepalive feature of pymemcache (dogpile.cache) cache backend.\n\nDepends-On: https://review.opendev.org/803716\nChange-Id: I7deed1bca1ce2f715967e966d4b0cedc1f227027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/60495a42c9e46144dc3625cef6ede24e18245011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/60495a42c9e46144dc3625cef6ede24e18245011"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
