)]}'
{"id":"openstack%2Fpuppet-oslo~812766","triplet_id":"openstack%2Fpuppet-oslo~master~Ic07fe61a530982ae6ae972f52775f5aec51b3134","project":"openstack/puppet-oslo","branch":"master","topic":"oslo-limit","hashtags":[],"change_id":"Ic07fe61a530982ae6ae972f52775f5aec51b3134","subject":"Add support for oslo.limit library parameters","status":"MERGED","created":"2021-10-07 01:47:23.000000000","updated":"2021-10-09 03:20:43.000000000","submitted":"2021-10-09 03:19:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812766-oslo-limit","meta_rev_id":"8a7ce6eeb44c022e2b48a61fe14212055ef7f9f1","_number":812766,"virtual_id_number":812766,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-09 03:19:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2021-10-08 10:34:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2021-10-09 01:16:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-10-07 21:49:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-10-08 19:28:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2021-10-09 01:16:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 03:53:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-07 21:49:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-10-08 10:34:45.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2021-10-08 19:28:42.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-10-09 01:16:39.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"e6551c5dcfc6645b38c05e06b676dbb10df79f49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-07 01:47:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84082a164e440e11f1e9c2e69cbc599b8037b84f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-07 01:51:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"29359ca4fe3f3ae99484510abdc5da1cf14d091f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-07 02:08:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ebd5af45e1de5064c94080453b97ad2042b4a5c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 03:53:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34ee3c704b464002aea7177a86516d27 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8370c75eb904bf1a239bacf05378b34 : SUCCESS in 3m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0667ae5d1b5464c86863deec9d614f0 : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee1a60fd090b4129a1531fa1d1b063a6 : SUCCESS in 11m 42s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/39240973a9cf4292a506edee27ac92d6 : SUCCESS in 11m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2284ff65059b4f59b9d5cbac41592382 : SUCCESS in 7m 57s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/feb3e413149946cb976776058271a6b2 : SUCCESS in 8m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e37b8e789a88481a956d4894e8dcc433 : SUCCESS in 7m 56s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b41cc915d4b84df28665946d0e5b225e : FAILURE in 7m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/367eef94e8ce4d388bd04dc17bfae383 : FAILURE in 8m 52s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0d27faab3d9448a9e2a2d6e97e107dd : FAILURE in 8m 51s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28208bc9ab4842a1b6ecc28e282c70a7 : FAILURE in 8m 19s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a56de612435b44e68155843d4ba83028 : SUCCESS in 23m 20s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/be3c5fa58467490c99053c7aafc563bf : SUCCESS in 1h 11m 10s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/54fe2aa3d6cb426bba398df8cf2fde73 : SUCCESS in 1h 37m 46s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7549e271e386461cbc16f8f66c59e8fa : SUCCESS in 1h 19m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/24f01f97120443c787583f7b3f0c53a8 : SUCCESS in 56m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0da6f1374fec4ad59c9a9be719b52157 : SUCCESS in 4m 15s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c374d4f71b03461eba97b6e048d6293f : SUCCESS in 1h 34m 40s","accounts_in_message":[],"_revision_number":3},{"id":"05c9bfd9b884835320c3598f47954ec49b6bcd79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-07 04:26:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8d0732a281a2ab03e08e24d191241dcdbd025e17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 06:08:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba19a5fe619c45b0938838a99ac96e94 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d918df19c7749c5bfd698df83165bb9 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c44993caa2f24976884ff7a29554efb6 : SUCCESS in 4m 28s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6727b04724544d56999728d2d21f3bed : SUCCESS in 13m 28s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/93931a9a808e4d0592dceb8a712e8a02 : SUCCESS in 10m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5189aea8438476fa9405533b0c36aa7 : SUCCESS in 10m 08s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1bb9bbaacf54438aeb817d5a9854268 : SUCCESS in 8m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7577201c2703476aabd7d826c54a3b14 : SUCCESS in 10m 09s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2976658a38f4450dbe236fea6b6ea88e : FAILURE in 7m 48s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/498335c5df064d85b663538667738cee : FAILURE in 6m 40s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fa70aa12b99424ba17200b8129f4c13 : FAILURE in 10m 27s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c59d565a64974163bdc2d2f38629afb0 : FAILURE in 7m 03s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/438ca1aa3b104e90ab0c4ab4a8c01c21 : SUCCESS in 22m 33s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ef8d524f7da4db789ae9bf7f1e1604c : SUCCESS in 1h 17m 49s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2cc8c59e928240b9ae714c04cb9bccaf : SUCCESS in 1h 26m 41s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b02ad14b7ce3415e92185ec50afe2b71 : SUCCESS in 1h 22m 42s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7db4db121ba4865aedc960644b8ca7d : SUCCESS in 1h 03m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65ff6c9192cd4aeda6200feab26f8c58 : SUCCESS in 4m 43s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e4cdb057085245d993b8cd459b33fa28 : SUCCESS in 1h 38m 21s","accounts_in_message":[],"_revision_number":4},{"id":"451ef5a9753a393adbf912e142e1825b4bdb434e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-07 21:49:21.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1a00c2c0623a3f0a4fcdcee9c37a5d8782bfb83d","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-10-08 10:34:45.000000000","message":"Patch Set 4: Code-Review+1\n\nThanks for this change..","accounts_in_message":[],"_revision_number":4},{"id":"22aa6d4780942f57d2b58e6587b096172f9bd0c3","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-10-08 19:28:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0b54f56a8c188a6c4f742d9ab5e7ec5b44498ce3","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-10-09 01:16:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"828c24095afce21680d00774c97a4bb9c1470425","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 01:18:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ed7c125190726b88870704b228373a9b29e464ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 03:19:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3493c8b7e224c9fb390f0f374d6f7c6 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/690a34c155ee42f68722f46f4662586a : SUCCESS in 3m 33s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c227fbae63984570bc371f59afb6701e : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f34a06d2ea540309255fefb3cf3e6ea : SUCCESS in 10m 28s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/44f0aa03ad7a429fbf2a47a518029b98 : SUCCESS in 9m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3df4c84e0874ac19de59ce12284e609 : SUCCESS in 7m 43s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee552de3172a4884aeb65c6b00c605e9 : SUCCESS in 7m 52s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e9359d2e1114be4aaa9e26e0c07b09c : SUCCESS in 1h 21m 02s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7059a5bff88a44418ef02bbd3ca88df9 : SUCCESS in 1h 26m 04s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed70dcd24df1473bb040b125d949f59c : SUCCESS in 1h 19m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0f849c4c6a146468c5ac0fca8685bec : SUCCESS in 58m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7791e4a6c4c4c8f93c35a1b205d45c6 : SUCCESS in 4m 10s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bf3e61c71b1c41ae9b6a4f473cca85df : SUCCESS in 1h 57m 20s","accounts_in_message":[],"_revision_number":4},{"id":"6e6ee83a9bfaeb77cee2a2fc3110ec6529b559b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 03:19:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"8a7ce6eeb44c022e2b48a61fe14212055ef7f9f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 03:20:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49d3b2ee19624a3884acd0b26e1b69d1 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f1ed9f4023efbc0d56fd0c6704a78b7ab1b8fb2d","revisions":{"3f1bdd2897b9715301cb7b0c94d024c4ce86e54f":{"kind":"REWORK","_number":1,"created":"2021-10-07 01:47:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/812766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/66/812766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/1"}}},"commit":{"parents":[{"commit":"1d71aa27977f0a8f6a5ad71e6876ef341a9132f4","subject":"Add information technology in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/1d71aa27977f0a8f6a5ad71e6876ef341a9132f4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-07 01:45:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-07 01:47:00.000000000","tz":540},"subject":"Add support for oslo.limit library parameters","message":"Add support for oslo.limit library parameters\n\nThis change introduces the new oslo::limit resource type, to manage\nparameters of the oslo.limit library[1]. This library is used to\nimplement the unified quota feature, and is used by Glance at the time\nof writing.\n\n[1] https://opendev.org/openstack/oslo.limit\n\nChange-Id: Ic07fe61a530982ae6ae972f52775f5aec51b3134\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3f1bdd2897b9715301cb7b0c94d024c4ce86e54f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3f1bdd2897b9715301cb7b0c94d024c4ce86e54f"}]},"branch":"refs/heads/master"},"a0f1542344c59d9bff9c0888f2ec98103c65a655":{"kind":"REWORK","_number":2,"created":"2021-10-07 01:51:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/812766/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/66/812766/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/2"}}},"commit":{"parents":[{"commit":"1d71aa27977f0a8f6a5ad71e6876ef341a9132f4","subject":"Add information technology in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/1d71aa27977f0a8f6a5ad71e6876ef341a9132f4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-07 01:45:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-07 01:51:30.000000000","tz":540},"subject":"Add support for oslo.limit library parameters","message":"Add support for oslo.limit library parameters\n\nThis change introduces the new oslo::limit resource type, to manage\nparameters of the oslo.limit library[1]. This library is used to\nimplement the unified quota feature, and is used by Glance at the time\nof writing.\n\n[1] https://opendev.org/openstack/oslo.limit\n\nChange-Id: Ic07fe61a530982ae6ae972f52775f5aec51b3134\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/a0f1542344c59d9bff9c0888f2ec98103c65a655"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/a0f1542344c59d9bff9c0888f2ec98103c65a655"}]},"branch":"refs/heads/master"},"f70c87b70d33b21fe6af90260cd76d71086d92fb":{"kind":"REWORK","_number":3,"created":"2021-10-07 02:08:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/812766/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/66/812766/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/3"}}},"commit":{"parents":[{"commit":"1d71aa27977f0a8f6a5ad71e6876ef341a9132f4","subject":"Add information technology in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/1d71aa27977f0a8f6a5ad71e6876ef341a9132f4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-07 01:45:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-07 02:07:53.000000000","tz":540},"subject":"Add support for oslo.limit library parameters","message":"Add support for oslo.limit library parameters\n\nThis change introduces the new oslo::limit resource type, to manage\nparameters of the oslo.limit library[1]. This library is used to\nimplement the unified quota feature, and is used by Glance at the time\nof writing.\n\n[1] https://opendev.org/openstack/oslo.limit\n\nChange-Id: Ic07fe61a530982ae6ae972f52775f5aec51b3134\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f70c87b70d33b21fe6af90260cd76d71086d92fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f70c87b70d33b21fe6af90260cd76d71086d92fb"}]},"branch":"refs/heads/master"},"f1ed9f4023efbc0d56fd0c6704a78b7ab1b8fb2d":{"kind":"REWORK","_number":4,"created":"2021-10-07 04:26:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/812766/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/66/812766/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/66/812766/4"}}},"commit":{"parents":[{"commit":"1d71aa27977f0a8f6a5ad71e6876ef341a9132f4","subject":"Add information technology in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/1d71aa27977f0a8f6a5ad71e6876ef341a9132f4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-07 01:45:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-07 04:26:21.000000000","tz":540},"subject":"Add support for oslo.limit library parameters","message":"Add support for oslo.limit library parameters\n\nThis change introduces the new oslo::limit resource type, to manage\nparameters of the oslo.limit library[1]. This library is used to\nimplement the unified quota feature, and is used by Glance at the time\nof writing.\n\n[1] https://opendev.org/openstack/oslo.limit\n\nChange-Id: Ic07fe61a530982ae6ae972f52775f5aec51b3134\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f1ed9f4023efbc0d56fd0c6704a78b7ab1b8fb2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f1ed9f4023efbc0d56fd0c6704a78b7ab1b8fb2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
