)]}'
{"id":"openstack%2Fpuppet-oslo~816294","triplet_id":"openstack%2Fpuppet-oslo~master~Ida5525f6d66145cbadc7ab5407b559b12b101a76","project":"openstack/puppet-oslo","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-03 13:39:31.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"Ida5525f6d66145cbadc7ab5407b559b12b101a76","subject":"Remove deprecated oslo_config","status":"MERGED","created":"2021-11-02 04:15:06.000000000","updated":"2021-11-03 17:43:07.000000000","submitted":"2021-11-03 17:40:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816294","meta_rev_id":"5018082a3f59b2940564da6ea8975a4af00a3abe","_number":816294,"virtual_id_number":816294,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-03 17:40:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-11-02 18:23:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-11-02 19:17:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-11-02 19:17:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-02 06:34:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-02 18:23:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-11-02 19:17:53.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"6b86bedf35b03e6a7291552075c517be26f69635","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 04:15:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a31a9bac015ebaebe6e3609169e5b8aa832cb9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 06:34:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/575d578580f34017a37f0ffaf9ad88eb : SUCCESS in 3m 14s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e632508be9c746a6a75dd97f8b4cc661 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0030e81448c6437e9def561b18149f7c : SUCCESS in 3m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02714e0f65c74c619f1ab6fb6851112d : SUCCESS in 3m 48s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3bea89b5f043438eab067f3378a769d4 : SUCCESS in 10m 07s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/32eb19893df2410abf3409e5b40135fb : SUCCESS in 9m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbfab07505024dbaa77e81f7da062e6a : SUCCESS in 6m 56s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/445079d403a0444ba4248a5373b9002d : SUCCESS in 7m 49s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c85abca3b4984c2b9b8d5035eeaaa9d7 : SUCCESS in 16m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc87541bae7f4113a08205a9535a4d3d : FAILURE in 1h 10m 12s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e75c7da0e63a404e8faf90fe1e1e44ff : FAILURE in 1h 16m 23s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3a065c43c514cf4a62f38a2e6f02994 : FAILURE in 25m 07s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecefd71be26641ed97d1b8698b8abb42 : SUCCESS in 1h 08m 14s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a4f129e0cf7456d860dd71c747c8d40 : SUCCESS in 23m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a93899defea74848ac7e1387117a3f5d : SUCCESS in 1h 11m 46s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ad1f5808e164905819ae37151835de3 : SUCCESS in 1h 18m 18s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/41e47d9c711b4792979c0ddb55439e66 : SUCCESS in 1h 27m 42s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1975490507274ffd926029022fe76254 : SUCCESS in 52m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/770d7be9e30845f18b65ed700ed347ff : SUCCESS in 4m 00s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/21fe15af47e541a48dac3ee2e2da8fb1 : SUCCESS in 2h 12m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4ebfbd18a6d3888e58b72ee808cf43401750dad0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-02 18:23:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7dd4abb4cae23523be17e74df37568731c8d9498","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-02 19:17:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73e50833bf050da84af00d46d8636c6a91140575","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-03 09:50:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8883019a703e6483910c95126b65b9311dd01569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 11:56:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/518faffdbf4d4f56a37fdf4659426e76 : SUCCESS in 4m 38s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93361f558335487fb05c483b80054f18 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f3a3bfc1b394800b1cf8c0ee9096992 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98f24a6961514c4b9fa54d03d35a371a : SUCCESS in 4m 29s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e9161462d3b414bb888a959b736fd16 : SUCCESS in 10m 55s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/948e9f111ad54872ad37705d3a6f4c73 : SUCCESS in 12m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ea344f8d4e54daa9cf1c816cae67b94 : SUCCESS in 8m 57s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c79d4fb990274de7a59ced58afe16887 : SUCCESS in 9m 31s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50f894f0b7ac420c832c64f6cde35689 : SUCCESS in 10m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e3798d70e014f67a8b8a0be11bf9927 : FAILURE in 37m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04ceeb476e0548869c754343354ea14d : FAILURE in 1h 29m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c91b3a00f47a40e882a9436f655e0986 : FAILURE in 1h 03m 31s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3ccaa14369b45dbb7f3d233104d22c5 : FAILURE in 50m 12s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f2840f70eaf4725924dc95192d16264 : SUCCESS in 23m 57s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e206fd6f6146435bbc6d5c56edcdf100 : SUCCESS in 1h 07m 30s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5aee7459f234dab9c62d49b005302c3 : SUCCESS in 1h 54m 31s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba3b9a44292f4176b2af4c040c303add : SUCCESS in 1h 51m 15s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f6d3b2d0cac5460cb6e432b7814b5d74 : SUCCESS in 59m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba4045324be9449987541b423e6962cb : SUCCESS in 6m 22s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/711cf3bf30b74dc39408b8dcc1791475 : SUCCESS in 1h 57m 30s","accounts_in_message":[],"_revision_number":1},{"id":"41f95d565654f8d8df9bdb60f22f6969ba2ed657","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 11:58:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c78bd78f59f605b02bfbaa5f9c36955e4574fba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 13:38:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/eb7d82aadfcb437bbc5e21d9d5c247ea : SUCCESS in 3m 19s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b84fe47c878742e7868c2d0aa5d9b51b : SUCCESS in 4m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b52404a680647d093bd688661aac681 : SUCCESS in 4m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e599fdb426b64b529d713799899422f2 : SUCCESS in 4m 22s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/098bc55237bf49d69ff08c197088c699 : SUCCESS in 13m 07s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f4759809ac84e5ca4c388e1ac07ca59 : SUCCESS in 11m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e13a877d3d0d4f8ab8f60533e7a641df : SUCCESS in 11m 19s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52070162a7fb4bbab3a7f426c7285139 : SUCCESS in 9m 02s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/727103b7ed954d8a8833514c28e8df7c : FAILURE in 1h 17m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bbf4bde753ae44b69affe5bced340e71 : SUCCESS in 1h 28m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/29937ccca9a94e5dafedddc27c54d8fe : SUCCESS in 1h 16m 52s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b59fa4b6ed5497e91788fb04aead1a9 : SUCCESS in 1h 13m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/717cea8969474cb0b2a73c675c55bd06 : SUCCESS in 5m 06s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/465bae78aeba47a9bdb914a3d3841bdc : SUCCESS in 1h 33m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d0fa071d2cea3481f569834496f847e62d7aae36","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-03 13:39:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95207437b1d4b86d54d9cda87bce40274b117582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 15:29:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/0abdbdc887f84f608580cbcd32cf00b3 : SUCCESS in 3m 11s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c204352116242a8a2fd4e2a07734db6 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88563e7dbf72456a8f09c16705174cd6 : SUCCESS in 4m 12s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/850313f9fc794a35b7a0fa6372e0e0e9 : SUCCESS in 4m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdc4f4f69d25411d984c71724385ea9c : SUCCESS in 12m 17s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f265791bc104afd81872187b382d707 : SUCCESS in 10m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af99aa8fad1d413f9332c03d8dd100a7 : SUCCESS in 9m 37s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad37c6899c8643d7b3ee82a9e1da76e2 : SUCCESS in 12m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/822f2fe1066a4ed299d68bfff7c95434 : SUCCESS in 8m 53s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56ee30cfef204f2f90e52462524ed5d9 : FAILURE in 1h 03m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4f5199dcadb40cebbc65cfbb604bcba : FAILURE in 1h 13m 12s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb5e935bec7449fa81a7c2f4dba70f4a : FAILURE in 40m 47s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eecd0937739a44efa0e9ee5b946b4572 : FAILURE in 38m 34s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8523fb12deb4482e8b13878b838693dc : SUCCESS in 23m 26s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd9cff5965324303913a78b49e982e50 : SUCCESS in 1h 14m 05s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/325c1762b30347de8f48d1deb00f13fa : SUCCESS in 1h 32m 05s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a05d9cf28720442d8a56c56187309a43 : SUCCESS in 1h 15m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3dd90677dddd4b27834465f52e0606d9 : SUCCESS in 57m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53d35a0da2944154a76159172469a252 : SUCCESS in 4m 38s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c6b913d199ea496d93d9949b4af18ac5 : SUCCESS in 1h 46m 13s","accounts_in_message":[],"_revision_number":1},{"id":"fc3ffb8bc43fbd8a7db4201f3fdb3f16555778c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 15:30:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a88de1e5d1821a11b86af659dec95b16d880a8ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 17:40:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/012c142534c241fe8abd856512c27231 : SUCCESS in 3m 32s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be4c2f883b4042cf90763bde0cfcb313 : SUCCESS in 5m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f6455808f7d41cea48ec40084dbd9e0 : SUCCESS in 4m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/057d350ffe8f44b69e63a98113aed855 : SUCCESS in 5m 03s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b19ba857fa2f45f185330bcd688d6fab : SUCCESS in 13m 07s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5d8a45d90764837a6da3318b4d0a56d : SUCCESS in 14m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb240918d13e48b2ae943997e73c0428 : SUCCESS in 11m 13s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/124d2564769349a9897b2a832a33acf8 : SUCCESS in 10m 24s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d61e2b8a7faa4f41a0b20bcbd012f745 : SUCCESS in 1h 20m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0d7bfcc86ce4cddbbb111b2ed9da859 : SUCCESS in 1h 27m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2047636182a74c85969a485d5d545461 : SUCCESS in 1h 23m 40s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5f2a1209f774aee8ff8f0506094d2a9 : SUCCESS in 1h 07m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ba79cf542b14985906df8d912ad61d2 : SUCCESS in 5m 25s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ebdf7ea878eb4a50bf5eab8f7c9a7371 : SUCCESS in 2h 04m 56s","accounts_in_message":[],"_revision_number":1},{"id":"94f7c8fcd4b7077dc83ff59aab7031df76e0d131","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 17:40:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5018082a3f59b2940564da6ea8975a4af00a3abe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 17:43:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57d45c7a808c49bdae72baa7af4d3654 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"056a1db5782709855b02bb47769a06457b03bae4","revisions":{"056a1db5782709855b02bb47769a06457b03bae4":{"kind":"REWORK","_number":1,"created":"2021-11-02 04:15:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/816294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/94/816294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/94/816294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/94/816294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/94/816294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/94/816294/1"}}},"commit":{"parents":[{"commit":"88a7885e0cb215cc37d0fdd55fb9de6a5e7967ba","subject":"Bump version for the start of Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/88a7885e0cb215cc37d0fdd55fb9de6a5e7967ba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 04:14:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 04:14:28.000000000","tz":540},"subject":"Remove deprecated oslo_config","message":"Remove deprecated oslo_config\n\n... because it was deprecated a while ago[1], and has not been used\nactually.\n\n[1] 6cd2bc408d55c7bdaf43013e1a2c0a1c4fd1fa9e\n\nChange-Id: Ida5525f6d66145cbadc7ab5407b559b12b101a76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/056a1db5782709855b02bb47769a06457b03bae4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/056a1db5782709855b02bb47769a06457b03bae4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
