)]}'
{"id":"openstack%2Fpuppet-oslo~819636","triplet_id":"openstack%2Fpuppet-oslo~master~I5b6431e93e3359cedb07192579258f9bc24b37c9","project":"openstack/puppet-oslo","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-07 23:00:34.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I5b6431e93e3359cedb07192579258f9bc24b37c9","subject":"[oslo_limit] password should be secret","status":"MERGED","created":"2021-11-29 13:04:08.000000000","updated":"2021-12-08 03:22:09.000000000","submitted":"2021-12-08 03:22:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819636","meta_rev_id":"a0e0267a9858d3964550c28e2d2642a7655d4195","_number":819636,"virtual_id_number":819636,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-08 03:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-30 09:02:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-12-07 16:04:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-12-07 16:04:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-29 15:04:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-30 09:02:31.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-12-07 16:04:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"fb77c0660e4f772b5de8c0af506b95eb8a4be9ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 13:04:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0501809ca8f8765cedb5948d3b11700ccebc2101","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 15:04:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48a12763ed674e58838a72c2e40e24bc : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80e0981ca8864003a9e3902036351b00 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cda559a47bc42bdbf56933749294354 : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8bcd41c03c0940da8288b42a53e14154 : FAILURE in 12m 38s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ec2fe6285c44b5d8c1e4b686feb22b3 : FAILURE in 11m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d29f0933a04457b848f899f3e5cb096 : FAILURE in 8m 51s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c4cb00ddcea427893ff62aa06832c73 : FAILURE in 8m 08s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb6eb32cbb1a4c8cbb251c16183ef55a : FAILURE in 10m 50s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ca113c779f94c448b2f400beeac2ced : FAILURE in 40m 46s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/530655967e814cc180f079c0d3f56f24 : FAILURE in 37m 57s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16bdea96f44b42ec89e5d42f97d9c30e : FAILURE in 32m 01s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54b48dd116984cf0bfb138c563f1e53a : FAILURE in 26m 17s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/de7edb6917a84aee9efc435f8c16d19a : SUCCESS in 24m 19s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c9a7e4b6ff94285b8a55df2a26c7f01 : SUCCESS in 1h 07m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7bf436a5282f4a0888b926cbea9b3fbc : SUCCESS in 1h 28m 22s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/871a2962d44244bd9e6977649711f155 : SUCCESS in 1h 22m 49s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/28cb11ffe10b44a8bebc6e42da9ae228 : SUCCESS in 1h 02m 36s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f637eaf3c1a478c83edb50b6dbaf767 : SUCCESS in 18m 46s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/06bcee2ce3634c45a7cc85011ff92e05 : SUCCESS in 56m 56s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/490200319e734871b3222801218dd2da : SUCCESS in 55m 26s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c897fb5f66dc470abfdb4f0089bc5f19 : SUCCESS in 1h 10m 31s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c8320d6b2da4d12983d5e6123b785be : SUCCESS in 42m 59s (non-voting)\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0f25ea8b40e147429c012f3938aed34c : SUCCESS in 1h 56m 46s","accounts_in_message":[],"_revision_number":1},{"id":"40351b087b12fc5f743ea2c8956744a5d0f31a34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 15:34:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2fb5f2facf8380c1df2db4ab8354082da3592c97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 18:09:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d2f6ac8ee4044a2a70bf03f2cd6543d : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52289ea13c7d4be885c9867e18f99697 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcb99d8fdafd4d25978df5d701b5944a : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/942c9faf37bb48eba8233c97e930ca61 : SUCCESS in 10m 58s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/595843c9c76d4f7ea91f705c94bdfbeb : SUCCESS in 12m 07s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0de395ca13eb4a5897970785e4d8f219 : SUCCESS in 7m 57s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9369a55c43342518861ebd553138b2e : SUCCESS in 7m 28s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3296dfa57f24e8dbc4793a327799d5b : SUCCESS in 10m 08s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a255a55ee944495a97e9fd6c25bfa9c1 : FAILURE in 31m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/691c28eeed074082882635c7309bebf1 : FAILURE in 1h 03m 36s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd90bdd25d9f49929c8624ad1ddf252f : FAILURE in 31m 41s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/204943b1df11464db4a41d5a8790bde7 : FAILURE in 27m 24s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/13e507d67d49458e98b87ae9c9a872aa : SUCCESS in 20m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b45071e7da341af957b819b008e2a23 : FAILURE in 1h 00m 25s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/281ba455c9944eb3b8cfebf22063fa5f : SUCCESS in 1h 42m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc7d876c155f4196ba7ce6dbb30f573d : SUCCESS in 1h 32m 21s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2421905448de4287beb30874e999ce6a : SUCCESS in 55m 16s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/897fe144b90348029003bca2df711ef8 : SUCCESS in 18m 43s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/04a6ccbe0d6d4230b3716f7f93cc735c : SUCCESS in 49m 38s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe3eaf3cd5b9415c8548e758c44ba359 : SUCCESS in 56m 21s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c0151ff4d3847f999c57348c3902b0c : SUCCESS in 57m 29s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/31585bc7160646e1b2eccaea4c8a9a56 : SUCCESS in 48m 04s (non-voting)\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/abf7f45d90304f84b98aab62db9e8946 : SUCCESS in 2h 27m 44s","accounts_in_message":[],"_revision_number":2},{"id":"8450816fdaff9dffd7ca16cd37963b48119d4971","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 23:03:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ac4421272aa36a64ae0798045ff9cabbe6723a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 01:08:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b532f27acbc448dabc2c2a39a93a2dc1 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57a680ad543a49738230b8b030108d19 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26fa78d2978d4212b26a62fce474d5af : SUCCESS in 4m 13s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/04088456a16c47e8aec3919b52342ecb : SUCCESS in 13m 20s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7af07ea16567434b917a0dbfec53deac : SUCCESS in 12m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b764303f8481419ea644ebd23217ff95 : SUCCESS in 9m 58s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8bdd2fc891b4b8287e9537bad18575f : SUCCESS in 7m 49s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b55357b46e3741d781ad338f7f621d85 : SUCCESS in 7m 50s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b462f6d1340e405b97765792d5e943ea : FAILURE in 30m 46s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55a19b087f9b4b93a3b8c92eff9fe3ff : FAILURE in 36m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3634522af941487f8b7496810acd0601 : FAILURE in 1h 06m 12s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b0b922a8b4b498e9a0c86ad6bbacbf5 : FAILURE in 26m 38s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb53b6883b6948a886ab92118cbe0f19 : SUCCESS in 21m 01s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/03b324f37b4348cc8fe8b0836326f554 : SUCCESS in 1h 15m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/91828bfd09d2413eaa2b422128f7de7e : SUCCESS in 1h 59m 16s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e08d54474e534537b56b562bccd3fbbe : SUCCESS in 1h 30m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8611aea0788546ae87ca8b9dbe30ceb8 : SUCCESS in 1h 03m 55s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b4e459b264f4141a4c7c1fce560432e : SUCCESS in 22m 49s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7afa17436c2945069f7ff9b3fb40239d : SUCCESS in 58m 46s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eefa7ca431d24a16a31107e767d95d33 : SUCCESS in 56m 45s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/20f496beebc841dbb77721deb851cad2 : SUCCESS in 57m 48s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/de7a40318af642d69a70723641c282fd : SUCCESS in 44m 48s (non-voting)\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/94ef682b0fc74865aefc79b9e2921dd5 : SUCCESS in 1h 56m 36s","accounts_in_message":[],"_revision_number":2},{"id":"8fe9f8fa7b2d39233f55ac92beb35672cd2957d5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-30 09:02:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dfc3edcd6101175f2bd272665785add62dfcab28","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-07 16:04:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"95ad01ddc78e30a7f975a9ecfaff400ae92dda38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 16:22:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"717703e470f1c50bf00494990dd7ba4797e1e586","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 18:58:09.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7eb2c58df8d643bb96ee0b564fe11b3b : SUCCESS in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01f77c1969fa495b8c95473d0bbf008c : SUCCESS in 3m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6f35bac2b1745cfbddf57cd9536bc30 : SUCCESS in 3m 41s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/38f4ef62e2c04e8b9b69b62496db082e : SUCCESS in 15m 36s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b49a05281944d62a11b4e1b210786c6 : SUCCESS in 9m 07s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/482a4a06bb284b1baffc31b9d091fd05 : SUCCESS in 11m 19s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48f08ebe453140b38b7a2d05888fac33 : SUCCESS in 8m 07s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e90c37e5c4c44d389bc5ed34273a6efe : FAILURE in 1h 47m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a1ed80e401f4bda82fa60cb908ed85d : SUCCESS in 1h 28m 22s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/90216a7b8f92475c823d8a1bcc644caa : SUCCESS in 1h 33m 49s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bcfad7fb33e84639be8e943573f85135 : SUCCESS in 1h 04m 38s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4634f17a77b949f0bda5c30afe7d0d4a : SUCCESS in 2h 34m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9bb0ea2ee31f6c894057ac1946ff6fa52c424c82","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 23:00:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bfbc743c9ddb0125ec99b5faff0f310c6bae33f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 01:13:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d576340e7694459ba92fc8da0dd58b0a : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/322d9c3489df473b93ca5420db26f92f : SUCCESS in 3m 33s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62d635c5940c48a4a6e36069124f97b4 : SUCCESS in 3m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/daf7fd2912e8458ab16a9f4222577019 : SUCCESS in 9m 46s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7aedf2cb1bd0460ead98f869ebec9cf1 : SUCCESS in 10m 13s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e92c229be49d46f3b082bf0a80b78dd3 : SUCCESS in 7m 30s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/173e6f5ecb3548888287538383c7beba : SUCCESS in 8m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d8ebc9ee8b24415be7b852ffa7dc753 : SUCCESS in 7m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc8f4151568b4fb0a316ef7bb756b5c7 : FAILURE in 32m 29s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbdff3fc37d24857a0588f9b197827d4 : FAILURE in 1h 17m 43s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a927378119f44fcdb9ab0e1f2472da5e : FAILURE in 1h 14m 11s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f354644e7d7640e09527c26ea28b990e : FAILURE in 1h 05m 55s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a6cf30165274976ab08e1b9d72c384b : SUCCESS in 21m 06s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0dd36f250b634858b19b72faa0f5b0f1 : SUCCESS in 1h 27m 36s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fafa69a073494c46aecbddbc1880bb95 : SUCCESS in 1h 27m 07s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ef4673ed016435fbdc33280eab0a289 : SUCCESS in 1h 21m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d9352c8c7ac4893a42be069209e1149 : SUCCESS in 1h 05m 02s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3612a23d7fae43c6a37e64d86e6b8a13 : SUCCESS in 19m 36s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2faa81f774de4463bd55a876546044d6 : SUCCESS in 50m 31s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/64941c7c3d484f6aa41678df8e210633 : SUCCESS in 51m 50s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/557567370c49494eb59593177feb6440 : SUCCESS in 1h 03m 12s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2bfeb10998c434998a44fe7a66eb1de : SUCCESS in 46m 02s (non-voting)\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5b49b310348c415eb02926af7bd15e45 : SUCCESS in 2h 07m 09s","accounts_in_message":[],"_revision_number":2},{"id":"34b735f53995499c9fee945f804969c1b996761e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 01:15:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0589e8ad21313d423eda829916b66758c6a12636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 03:22:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cac42ea6162488fb38a120396343617 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4d0539fb5884427ad6d477f435a1828 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e57579f14b54b10be88dc46cc0e98e4 : SUCCESS in 3m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2666b677689a452397d230dcae3d0602 : SUCCESS in 10m 56s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/86141afb5d6141148e3c7ea3408e4981 : SUCCESS in 10m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7e2d23c0c294a69bc4b85d3bdcb9b34 : SUCCESS in 7m 18s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f7c96d413144655ad1040a212a9be3f : SUCCESS in 9m 09s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/395535e5c60d4610b508986a7621c353 : SUCCESS in 1h 25m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c76d04d58e4d454398c0395ec26aecc4 : SUCCESS in 1h 29m 52s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b37a4b7ed40a472fa89f152fc3660670 : SUCCESS in 2h 01m 43s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/df6242cb6af245e6b59f503f2e0b079d : SUCCESS in 1h 02m 55s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6628044528554b1a996eedc5ddae62e6 : SUCCESS in 1h 44m 52s","accounts_in_message":[],"_revision_number":2},{"id":"a0e0267a9858d3964550c28e2d2642a7655d4195","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 03:22:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9da6c6d9fcdc0af941554804af670debaf31870e","revisions":{"c4b2458123786bbdd8f26701076bd0a5388dbf96":{"kind":"REWORK","_number":1,"created":"2021-11-29 13:04:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/819636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/36/819636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/36/819636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/36/819636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/36/819636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/36/819636/1"}}},"commit":{"parents":[{"commit":"056a1db5782709855b02bb47769a06457b03bae4","subject":"Remove deprecated oslo_config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/056a1db5782709855b02bb47769a06457b03bae4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 13:03:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 13:03:58.000000000","tz":540},"subject":"[oslo_limit] password should be secret","message":"[oslo_limit] password should be secret\n\nChange-Id: I5b6431e93e3359cedb07192579258f9bc24b37c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c4b2458123786bbdd8f26701076bd0a5388dbf96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c4b2458123786bbdd8f26701076bd0a5388dbf96"}]},"branch":"refs/heads/master"},"9da6c6d9fcdc0af941554804af670debaf31870e":{"kind":"REWORK","_number":2,"created":"2021-11-29 15:34:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/819636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/36/819636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/36/819636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/36/819636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/36/819636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/36/819636/2"}}},"commit":{"parents":[{"commit":"056a1db5782709855b02bb47769a06457b03bae4","subject":"Remove deprecated oslo_config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/056a1db5782709855b02bb47769a06457b03bae4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 13:03:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 15:33:57.000000000","tz":540},"subject":"[oslo_limit] password should be secret","message":"[oslo_limit] password should be secret\n\nChange-Id: I5b6431e93e3359cedb07192579258f9bc24b37c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9da6c6d9fcdc0af941554804af670debaf31870e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9da6c6d9fcdc0af941554804af670debaf31870e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
