)]}'
{"id":"openstack%2Fpuppet-oslo~819975","triplet_id":"openstack%2Fpuppet-oslo~master~I681987e8ad3781a06897817fc8432abc69aa474f","project":"openstack/puppet-oslo","branch":"master","topic":"prep-release-yoga-m2","attention_set":{},"removed_from_attention_set":{"9414":{"account":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"last_update":"2022-01-07 14:37:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I681987e8ad3781a06897817fc8432abc69aa474f","subject":"Prepare Yoga M2","status":"MERGED","created":"2021-12-01 06:43:42.000000000","updated":"2022-01-07 14:37:02.000000000","submitted":"2022-01-07 14:37:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819975-prep-release-yoga-m2","meta_rev_id":"e0f2c1ca7d671e3684f5f688380cbf4e9e25d3ee","_number":819975,"virtual_id_number":819975,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 14:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-01-07 14:22:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-01-07 14:37:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-01-07 14:22:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-01 06:53:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-13 15:37:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-07 14:22:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"f03a96433dc9ee9c04264fec41d5bb859dc49b26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-12-01 06:43:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef4a7bd81330be4ea0fc1d7cb27461bdd390c5ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 06:53:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/122abc6800f24cb2b018a9f6892e4c07 : SUCCESS in 2m 34s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67979bcbe9f34245a56320380a14f659 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"cdd3ff20ccd42ced171b78a289bf814109e9ec1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-12-03 08:56:15.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5ab85808218dc6967061e1b73f44891eafe6e43c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 09:06:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/9dfd4711dbcc43cd89a2fe4311ccbbda : SUCCESS in 2m 39s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c57cca77f664597a809a81ddbf5b372 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":2},{"id":"caaa9f04c7e8cbdcb8273bd944f54e02115ad380","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-13 15:37:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dc142aad493e32912da199c270defa6259aee3de","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-03 02:42:27.000000000","message":"Topic prep-release-yoga-m1 removed","accounts_in_message":[],"_revision_number":2},{"id":"9db902b7cca8e0f5da54632f9d251ca15a09cb62","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-03 02:42:29.000000000","message":"Topic set to prep-release-yoga-m2","accounts_in_message":[],"_revision_number":2},{"id":"59d5dfde520871bfdbc7aa7df484584d587a6635","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-07 14:22:29.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"20272f4f4cc741bff848ba80868ef3dac6ade2d3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-07 14:22:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c95c3750ab7edfe16d9f3a12edf84f141aeee2f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:30:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/c24eec4818394a11b39212d47c0d6159 : SUCCESS in 2m 44s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f9f170c95134a7ea81574bf633f372a : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3},{"id":"30c426651c00e5549a5ec4f8262205d6eaa97b61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:31:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7e4b0ab48b67ba0587ce967c35d20f993c6704e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:37:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/659cd7a58c4b4596adc0692c947260c8 : SUCCESS in 2m 57s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e65d27cf4b9f4deaacbc7900d2f01b74 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":3},{"id":"6c4f06ac697fae5fdf6cd319e11de3c2c0afcec8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:37:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a9e9682881a4a039c7b973c0618c404a3626b044","revisions":{"4e2e7fe8f6961d122986a1dc165009004c3d1318":{"kind":"REWORK","_number":1,"created":"2021-12-01 06:43:42.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/75/819975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/75/819975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/75/819975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/75/819975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/75/819975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/75/819975/1"}}},"commit":{"parents":[{"commit":"056a1db5782709855b02bb47769a06457b03bae4","subject":"Remove deprecated oslo_config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/056a1db5782709855b02bb47769a06457b03bae4"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:52.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:52.000000000","tz":480},"subject":"Prepare Yoga M1","message":"Prepare Yoga M1\n\nUpdate the version metadata for Yoga milestone M1\n\nChange-Id: I681987e8ad3781a06897817fc8432abc69aa474f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/4e2e7fe8f6961d122986a1dc165009004c3d1318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/4e2e7fe8f6961d122986a1dc165009004c3d1318"}]},"branch":"refs/heads/master"},"0c9ac1ac1a6f258b56c705c0d8bcb3d630f92312":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-03 08:56:15.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/75/819975/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/75/819975/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/75/819975/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/75/819975/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/75/819975/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/75/819975/2"}}},"commit":{"parents":[{"commit":"056a1db5782709855b02bb47769a06457b03bae4","subject":"Remove deprecated oslo_config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/056a1db5782709855b02bb47769a06457b03bae4"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:52.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-02 06:00:27.000000000","tz":480},"subject":"Prepare Yoga M2","message":"Prepare Yoga M2\n\nUpdate the version metadata for Yoga milestone M2\n\nChange-Id: I681987e8ad3781a06897817fc8432abc69aa474f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/0c9ac1ac1a6f258b56c705c0d8bcb3d630f92312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/0c9ac1ac1a6f258b56c705c0d8bcb3d630f92312"}]},"branch":"refs/heads/master"},"a9e9682881a4a039c7b973c0618c404a3626b044":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-07 14:22:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/819975/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/75/819975/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/75/819975/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/75/819975/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/75/819975/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/75/819975/3"}}},"commit":{"parents":[{"commit":"9da6c6d9fcdc0af941554804af670debaf31870e","subject":"[oslo_limit] password should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9da6c6d9fcdc0af941554804af670debaf31870e"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:52.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-07 14:22:29.000000000","tz":0},"subject":"Prepare Yoga M2","message":"Prepare Yoga M2\n\nUpdate the version metadata for Yoga milestone M2\n\nChange-Id: I681987e8ad3781a06897817fc8432abc69aa474f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/a9e9682881a4a039c7b973c0618c404a3626b044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/a9e9682881a4a039c7b973c0618c404a3626b044"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
