)]}'
{"id":"openstack%2Fpuppet-oslo~823629","triplet_id":"openstack%2Fpuppet-oslo~master~If4f1633c6dd7adf4b80c0a8cc83ddd3d025d099b","project":"openstack/puppet-oslo","branch":"master","topic":"system-scope-params","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-07 23:48:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4f1633c6dd7adf4b80c0a8cc83ddd3d025d099b","subject":"Accept system scope credentials for Unified Limits API","status":"MERGED","created":"2022-01-06 06:41:02.000000000","updated":"2022-01-07 23:50:06.000000000","submitted":"2022-01-07 23:48:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823629-system-scope-params","meta_rev_id":"bdb4d3963cf048fafa861d7c3cea27a1d28117c0","_number":823629,"virtual_id_number":823629,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 23:48:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-01-07 16:39:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-01-07 21:54:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-01-07 21:54:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 08:47:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-07 16:39:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-07 21:54:21.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"ac97dc39c0d2b4aae7b8aaae3e010efd27c3c369","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-06 06:41:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10e8c1e357e9abcf8368fb89e476be9448b39ccb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-06 06:46:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"df8516c183b31b9f6e0aa4f83b94ba3e0396d5b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 08:47:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44f0cb8c3e7b41b680c37737b9514ddf : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b21dae3f4d1c4396a1e10d098433f6ff : SUCCESS in 3m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbf3f623cf394e4981e16a4b44aedd25 : SUCCESS in 4m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7eb0e7dcba8348c38ff68dce493c8857 : FAILURE in 11m 20s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9fb039fe24db41ceadb444a285492464 : FAILURE in 13m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd966e0eb7474e31b6010d1a992849df : FAILURE in 8m 10s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4227f6a6ad68481aa78a2f1547fa42d0 : FAILURE in 10m 49s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55ba0a131b09461ca9d8855999932bbf : FAILURE in 12m 02s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69690ff443164d478a4ad712ec64e5f3 : FAILURE in 36m 53s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a7fc3ece9444f1b8e43a680c031d249 : FAILURE in 24m 28s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e31e47decf9f41c8b06ed5a358c15669 : FAILURE in 24m 13s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d034a4d389b641d18b8ad5fc80886ff6 : FAILURE in 26m 26s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4460f738ff474cd78564823681dd7da5 : SUCCESS in 26m 39s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f6707269979b40cfb3415640fd820d39 : SUCCESS in 1h 19m 27s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6748b5d05a5e4effbd2d8e01ed1c2672 : SUCCESS in 1h 30m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2bd2475a8644be1920dd2c739dc2a79 : SUCCESS in 1h 32m 14s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ed82628d90841729205e0e48ae65661 : SUCCESS in 1h 01m 33s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4c24001e20f4b8fbf93cdaaef764d8f : SUCCESS in 21m 32s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cd12e1649cc4873b73da05ff7712812 : SUCCESS in 53m 24s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/115840c3aa1640efa16419493f710b79 : SUCCESS in 1h 02m 13s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/310bd984a2a94e8497e267f7880c29ee : SUCCESS in 1h 03m 36s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a83efa80c0a1458aa58074ac6a66907b : SUCCESS in 43m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/585783401b67450ea5a0a659900b3c89 : SUCCESS in 3m 53s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9ed1a7ee8f1943c9a91ee753ff5c40d3 : SUCCESS in 1h 58m 00s","accounts_in_message":[],"_revision_number":2},{"id":"192f09e2c0856a76fd539afabc931bb45fb64a38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-06 08:53:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"86abed42e6b6239aae1754e2d6092c978b0f7148","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-06 08:54:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d17aabf5e8ec60d2015e0b4ba39c8a4ba8699ce8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 11:19:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bc7b11ec0644747a4a21190c389f8f5 : SUCCESS in 3m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9187431948cb482cb4a713eb8d2bb87a : SUCCESS in 3m 29s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5129b295d9064a48b4a490b3f69c1975 : SUCCESS in 3m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/cce7905669f84625bb780e0cef312b8c : FAILURE in 11m 28s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ad3b8721fbc4eceb94c70806acb472e : FAILURE in 10m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff528bbac13f4a76b9a2b583330ca1c7 : FAILURE in 7m 28s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ae17439331e47d28fefca7995ae1626 : FAILURE in 8m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/381b265d74234d8cb2ab04c1f7eba9fa : FAILURE in 7m 46s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b60aa2199ae8483e856796013eee8ba5 : FAILURE in 31m 15s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7101c4801ea848b495017fd56811160a : FAILURE in 27m 37s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84eb92db86d548d78f8304b4afde73ed : FAILURE in 24m 50s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/322e7d36ba63410f80286062f2f20a87 : FAILURE in 23m 53s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a46429e3ba9640b2836477cf9a41ac12 : SUCCESS in 24m 31s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c37067f07a14499cb4c56f889ece5904 : SUCCESS in 1h 17m 49s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/20bd5c6ace5341538c51efb8409ac058 : SUCCESS in 1h 18m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c2bb13609acb40f1939334f3ed233f08 : SUCCESS in 1h 09m 20s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb4a4bf28d2d45cd84e6e5c7985c1abb : SUCCESS in 59m 15s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/389181b2228a4a0aa3a3060773619d6c : SUCCESS in 19m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c67bdeae07e482f8e8b162c3be5f06b : SUCCESS in 58m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7437f0bdfcc4c2da9f3507c07e2654e : SUCCESS in 1h 02m 36s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c911676a7db4c129528988443c08d61 : SUCCESS in 1h 24m 27s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/050243d4294a4822894801ce86f16e27 : SUCCESS in 49m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b88721782e7b43fdb5f0472d666e6766 : SUCCESS in 4m 40s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/53adc223cf124c158b91eaf9460332bb : SUCCESS in 2h 17m 39s","accounts_in_message":[],"_revision_number":4},{"id":"adf7168581cf939464bd54cef8695faf363d72cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-06 11:33:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9b276cba20cf1a14db4a2031eaafb09907f1a706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 13:39:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73df3a58eedd4486b14f3e74d57afd36 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c003a1e522df4074bc3ba0b2ce3b60bd : SUCCESS in 3m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cc9f9c14b4a4f849680d4f6ae185890 : SUCCESS in 3m 43s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/abb661bf55704a83b90ecf5da5bd460e : FAILURE in 13m 15s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2868e1a7f2bd47f4aab86890d92beb8b : FAILURE in 11m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/123bb6d866644697aa2f6f486b1b407d : FAILURE in 9m 24s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67efc74dd11f4600a172d94077fecfcc : FAILURE in 8m 54s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c2a7879e6ec41beb2e4e2e10c08d43f : FAILURE in 9m 31s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94d3140d0b7c494a9fcdaf1b4c5b0135 : FAILURE in 29m 27s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aaa84b231bb64b11ae5b478eb7bb92d6 : FAILURE in 28m 43s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eee8bd03db9347f78033469761aab048 : FAILURE in 26m 49s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b145c8ca242040afbed405b61d1e6d99 : FAILURE in 25m 25s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a1c1afdacd1344e185ca1a890fb19fd4 : SUCCESS in 26m 58s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/de0b34be066247519b96c6a974a3db63 : SUCCESS in 1h 19m 12s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cec35a611c644c98bbb08332cccc2a44 : SUCCESS in 1h 37m 05s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c2a5f7b1200244a7856e409e5d5edfb4 : SUCCESS in 1h 26m 36s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/06dca59b7bbc46179a27cee8da32e768 : SUCCESS in 1h 11m 19s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/17d4111cb0174305b5573ac07aa9f558 : SUCCESS in 21m 19s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b73a80d32ce48a3af11f9334254d973 : SUCCESS in 59m 35s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/822cf7bb0ee348c487aad1f7b5eb6681 : SUCCESS in 53m 24s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5160b5bc50d745539ca30db9228c83f2 : SUCCESS in 56m 21s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/93dc8153e0b6448d9c1e6e15afebd894 : SUCCESS in 54m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9643c7d5738f40e3b42252893976be59 : SUCCESS in 3m 49s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6dd7911139474104847a507f24e08891 : SUCCESS in 1h 59m 10s","accounts_in_message":[],"_revision_number":5},{"id":"e7d6ee546aafede1c59fe638b32587b506e439c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-06 13:43:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"70a90d5242932e0c0af1ec8c7ff3033c085aebab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 16:02:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8a51a694a264f528cac854e8653eb89 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c078909543e74ae18aeb46083b3ef840 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4191495eccf748d2b14b683ee75527ef : SUCCESS in 4m 26s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/beaf010dc8374e12a306797ecf0fc8e4 : SUCCESS in 12m 32s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/174d63aed0d446729266423efbab6852 : SUCCESS in 11m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c451c6611284184bd41ab77470e7627 : SUCCESS in 8m 33s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdcc809394e04339b088e96d1ea2f93d : SUCCESS in 8m 37s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d89a85f8f7734e2baa79a716b91ddfc3 : SUCCESS in 9m 03s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86d00f560c864aadb987e3b3f4bf4313 : FAILURE in 26m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d2eec334cbb47f4bcbe3e6f58323867 : FAILURE in 27m 45s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb478ab98d4044388d6ea56e6a8ec396 : FAILURE in 28m 23s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/367487db51d24136aaf97b2e62e86556 : FAILURE in 20m 14s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/acf53d6678884b22b4a8d91fc0120e79 : SUCCESS in 25m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc1166fcc4444c23bb0a5d491c28a37f : SUCCESS in 1h 16m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/72d4e0b8fd1e4bf3a4a0fbb45ebd2a4e : SUCCESS in 1h 27m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a9ef4dda7714ff787dbd73418ea1c42 : SUCCESS in 1h 41m 34s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a05ec763f97344c49cf93d4d6b732112 : SUCCESS in 59m 34s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/35b8755479cc468ba05b6521f6e26fba : SUCCESS in 19m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/24153a900fda4ee68cfce1963e4c9ffd : SUCCESS in 52m 34s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3556a0f245f40858163859dc81c9a3f : SUCCESS in 57m 03s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ba1baa909bc40449466937bad47f418 : SUCCESS in 1h 02m 11s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2eb73e76f4d64b978f43da1d310251bc : SUCCESS in 48m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cce1eb7980f747efb9f4e06375b1bc46 : SUCCESS in 4m 38s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9788c9bef3ef49a4bb8ff873f3ba729d : SUCCESS in 2h 15m 21s","accounts_in_message":[],"_revision_number":6},{"id":"e605c505ad6740366bd4e68b69333a579d51ec3d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-07 16:39:37.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"434c03712e199453f2ec50bdf6db49385aa6d637","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-07 21:54:21.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"658ab36edb80ce0c4c95b6c555b30cd60742545b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 21:56:33.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c897deb1a14e4284c6108b50bfc6bab671219291","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:48:37.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41cc9ffa3b884d80803f4ad532cea464 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ebc0d87f6334f83a40ac7536f7e6368 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36ab382c7eb74ef7b07cb4d2fb2ad742 : SUCCESS in 3m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ac91e249b7c41c5a17162c39f478117 : SUCCESS in 12m 06s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a8bc0881cb94e80af16697adaf7a3f0 : SUCCESS in 10m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e28a17f650d4e619eae54b3ed3c764f : SUCCESS in 8m 39s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07ff09b5148f43b3a646c06ad43668e7 : SUCCESS in 8m 30s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e30c762b11543dc8828e502eccff95d : SUCCESS in 1h 07m 49s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e9e5145fd834548851c76d4d97e0446 : SUCCESS in 1h 25m 16s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/42d6520eebf44f42aa1d58e0c1c43b86 : SUCCESS in 1h 23m 20s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bef37b30e4a1470db70f70598ca9de4d : SUCCESS in 1h 00m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6686ba76f5444cd1823ad6fd09376cc6 : SUCCESS in 3m 57s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/578ccf8327ab43319219515ea2ef13ab : SUCCESS in 1h 50m 56s","accounts_in_message":[],"_revision_number":6},{"id":"c3b3eb88bd0156fc33d8fc1ca93ddcb5b773283f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:48:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"bdb4d3963cf048fafa861d7c3cea27a1d28117c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:50:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3ef08fbcae1468a9df776e9e6264e84 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"937fcf0644e36060c7e90dc4e7b708fc240b9c25","revisions":{"c51d46738f5d87bf0243942cd7b2b063b12e5326":{"kind":"REWORK","_number":1,"created":"2022-01-06 06:41:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/823629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/29/823629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/1"}}},"commit":{"parents":[{"commit":"9da6c6d9fcdc0af941554804af670debaf31870e","subject":"[oslo_limit] password should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9da6c6d9fcdc0af941554804af670debaf31870e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-06 06:38:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-06 06:38:02.000000000","tz":540},"subject":"Accept system scope credentials for limit API","message":"Accept system scope credentials for limit API\n\nThis change allows usage of system scope credentials in addition to\nproject scope credentials to use the limit API in Keystone for unified\nquota feature.\n\nChange-Id: If4f1633c6dd7adf4b80c0a8cc83ddd3d025d099b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c51d46738f5d87bf0243942cd7b2b063b12e5326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c51d46738f5d87bf0243942cd7b2b063b12e5326"}]},"branch":"refs/heads/master"},"3fcc5599d06a913ca7e57918d88f6c6a31d7707d":{"kind":"REWORK","_number":2,"created":"2022-01-06 06:46:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/823629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/29/823629/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/2"}}},"commit":{"parents":[{"commit":"9da6c6d9fcdc0af941554804af670debaf31870e","subject":"[oslo_limit] password should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9da6c6d9fcdc0af941554804af670debaf31870e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-06 06:38:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-06 06:45:57.000000000","tz":540},"subject":"Accept system scope credentials for Unified Limits API","message":"Accept system scope credentials for Unified Limits API\n\nThis change allows usage of system scope credentials in addition to\nproject scope credentials to use the Unified Limits API in Keystone.\n\nChange-Id: If4f1633c6dd7adf4b80c0a8cc83ddd3d025d099b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3fcc5599d06a913ca7e57918d88f6c6a31d7707d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3fcc5599d06a913ca7e57918d88f6c6a31d7707d"}]},"branch":"refs/heads/master"},"63cdd3c4fa790772c21da9df15ae2d1843217d82":{"kind":"REWORK","_number":3,"created":"2022-01-06 08:53:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/823629/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/29/823629/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/3"}}},"commit":{"parents":[{"commit":"9da6c6d9fcdc0af941554804af670debaf31870e","subject":"[oslo_limit] password should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9da6c6d9fcdc0af941554804af670debaf31870e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-06 06:38:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-06 08:53:30.000000000","tz":540},"subject":"Accept system scope credentials for Unified Limits API","message":"Accept system scope credentials for Unified Limits API\n\nThis change allows usage of system scope credentials in addition to\nproject scope credentials to use the Unified Limits API in Keystone.\n\nChange-Id: If4f1633c6dd7adf4b80c0a8cc83ddd3d025d099b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/63cdd3c4fa790772c21da9df15ae2d1843217d82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/63cdd3c4fa790772c21da9df15ae2d1843217d82"}]},"branch":"refs/heads/master"},"875f5511d8dbda0b3a22b8dc3254e2c6e57a5779":{"kind":"REWORK","_number":4,"created":"2022-01-06 08:54:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/823629/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/29/823629/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/4"}}},"commit":{"parents":[{"commit":"9da6c6d9fcdc0af941554804af670debaf31870e","subject":"[oslo_limit] password should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9da6c6d9fcdc0af941554804af670debaf31870e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-06 06:38:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-06 08:54:20.000000000","tz":540},"subject":"Accept system scope credentials for Unified Limits API","message":"Accept system scope credentials for Unified Limits API\n\nThis change allows usage of system scope credentials in addition to\nproject scope credentials to use the Unified Limits API in Keystone.\n\nChange-Id: If4f1633c6dd7adf4b80c0a8cc83ddd3d025d099b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/875f5511d8dbda0b3a22b8dc3254e2c6e57a5779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/875f5511d8dbda0b3a22b8dc3254e2c6e57a5779"}]},"branch":"refs/heads/master"},"4ebc65c4e775fd7200c70b5eddf4424922d1edf0":{"kind":"REWORK","_number":5,"created":"2022-01-06 11:33:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/823629/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/29/823629/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/5"}}},"commit":{"parents":[{"commit":"9da6c6d9fcdc0af941554804af670debaf31870e","subject":"[oslo_limit] password should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9da6c6d9fcdc0af941554804af670debaf31870e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-06 06:38:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-06 11:32:52.000000000","tz":540},"subject":"Accept system scope credentials for Unified Limits API","message":"Accept system scope credentials for Unified Limits API\n\nThis change allows usage of system scope credentials in addition to\nproject scope credentials to use the Unified Limits API in Keystone.\n\nChange-Id: If4f1633c6dd7adf4b80c0a8cc83ddd3d025d099b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/4ebc65c4e775fd7200c70b5eddf4424922d1edf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/4ebc65c4e775fd7200c70b5eddf4424922d1edf0"}]},"branch":"refs/heads/master"},"937fcf0644e36060c7e90dc4e7b708fc240b9c25":{"kind":"REWORK","_number":6,"created":"2022-01-06 13:43:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/823629/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/29/823629/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/29/823629/6"}}},"commit":{"parents":[{"commit":"9da6c6d9fcdc0af941554804af670debaf31870e","subject":"[oslo_limit] password should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9da6c6d9fcdc0af941554804af670debaf31870e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-06 06:38:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-06 13:43:04.000000000","tz":540},"subject":"Accept system scope credentials for Unified Limits API","message":"Accept system scope credentials for Unified Limits API\n\nThis change allows usage of system scope credentials in addition to\nproject scope credentials to use the Unified Limits API in Keystone.\n\nChange-Id: If4f1633c6dd7adf4b80c0a8cc83ddd3d025d099b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/937fcf0644e36060c7e90dc4e7b708fc240b9c25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/937fcf0644e36060c7e90dc4e7b708fc240b9c25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
