)]}'
{"id":"openstack%2Fpuppet-oslo~823883","triplet_id":"openstack%2Fpuppet-oslo~master~I3ab30adbae3e805aee36bd32854e5490b19b0097","project":"openstack/puppet-oslo","branch":"master","topic":"system-scope-params","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-10 21:14:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3ab30adbae3e805aee36bd32854e5490b19b0097","subject":"barbican_service_user: Accept system scope credential","status":"MERGED","created":"2022-01-08 03:16:25.000000000","updated":"2022-01-10 21:17:23.000000000","submitted":"2022-01-10 21:14:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823883-system-scope-params","meta_rev_id":"e1e8679fe83edd0ea92d76b1fdd3816225678492","_number":823883,"virtual_id_number":823883,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-10 21:14:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-01-10 08:03:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-01-10 18:49:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-01-10 18:49:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-08 06:15:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-10 08:03:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-01-10 18:49:20.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"08b01b92914f9ca9f7432cd222f3c3f2fc9c19e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-08 03:16:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65c2e92ff549020820c738d04d4c999f83f4c52f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-08 04:20:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"90d311b5193007579b0946fdc52999696810de30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 06:15:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fda019a7ad534e0ba91e4f91854d6c92 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bc312975ed742f68800eeb4b35e11cc : SUCCESS in 3m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f7b5614136448ae8e498d83bbb6ac7c : SUCCESS in 6m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fe42ad3c698408baad8a96b0182f7a3 : FAILURE in 13m 11s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f7836713c40f43b9b7b98d37092bb734 : FAILURE in 11m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef30ed2f497a4371a00ae4ead5da23b4 : FAILURE in 12m 07s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcf03a641b084f87a6ea85475385066f : FAILURE in 10m 18s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb0fda25db3a492fbf92d7d97b21af01 : FAILURE in 10m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a75bbe01f434f64b9ca556968c8946d : FAILURE in 26m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b765b0d8ddbd4e21ace4fcf969e44699 : FAILURE in 23m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08df2fddccae456ca8c85717ab5accbc : FAILURE in 21m 21s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d938c6739ea44d1a39bac1eee0ba703 : FAILURE in 22m 04s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f5f9b483b964bdc8c2e3a1250dc4f6e : SUCCESS in 21m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2f6a4c6d90f4f32baa811555a67c963 : SUCCESS in 59m 19s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4d3e86bf4b54e56845ba2159ae1ef55 : SUCCESS in 1h 53m 22s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f34f1f8a54b14999a012169b34b63017 : SUCCESS in 1h 03m 13s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3dfebd552e74f4fa118265382de8283 : SUCCESS in 1h 04m 14s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed33d29198ae4854b4fcecccb5ee5b38 : FAILURE in 10m 59s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9b21f1f27814cadb3265ef3a0194ad4 : SUCCESS in 55m 26s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e7fe0dbab5646a9af963ed7e86fd3c5 : SUCCESS in 44m 36s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f117c4b9eb1044bc87977966e77baf4e : SUCCESS in 1h 02m 22s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4eeb345d1514eaba483226266e4ca70 : SUCCESS in 39m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e537e61045547e5b06dd59f7cc23d7c : SUCCESS in 3m 33s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b8ae65c80d714d46a948de7736f97a87 : SUCCESS in 1h 44m 35s","accounts_in_message":[],"_revision_number":2},{"id":"b742533cea9f0daa45b5b1221a435456335666e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-08 06:26:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"894267b89e8c85f07c0525a70a6a09d48a6b49d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 08:37:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32fa8bbc0c8841fcbdac2ba04a9958c6 : SUCCESS in 7m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e46ef20f61fd49bfac8f8d93f534e85b : SUCCESS in 3m 05s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/009a8fba2c2e4efeab083deed8f1592c : SUCCESS in 2m 59s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/296158025f604c439cd929f964ba2165 : SUCCESS in 10m 17s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/fdfe73a881c642e885cbcc50e6e5a10e : SUCCESS in 10m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fcfc9b861244bdb85d4f8b151af7d3b : SUCCESS in 8m 11s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fea5bb9c42eb43a1808d7434694eac9c : SUCCESS in 8m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/938d3941d18c45f7a07d90093b28faf0 : SUCCESS in 7m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cbed46ea2b24af88833ea94eafa6319 : FAILURE in 23m 48s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0f01eda920d4a689837f56f505cfa0a : FAILURE in 23m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4727ec9cb1f842c697675cfe18118e4a : FAILURE in 21m 51s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac8f5231e25c41fe91d77bc26ea10118 : FAILURE in 21m 11s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/13a6f5e8c46d451691a9c3acceb9060d : SUCCESS in 22m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d56acf5d7c4e4cfb84e3e722eac70fea : SUCCESS in 1h 03m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6fea909505e549ef8dd7ca409bc7be8d : SUCCESS in 1h 04m 18s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/156e9ba32393435288c34e4f8e1c0019 : SUCCESS in 1h 07m 48s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e1e7e6c8a8f4cec8806464ac0f670a2 : SUCCESS in 51m 31s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd28c358645340d591676c372cdb00e7 : SUCCESS in 24m 18s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5269710f21d5461d95dd216e30ee1519 : SUCCESS in 42m 00s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/df6f82b53092419c93c6d640711b5df2 : SUCCESS in 57m 15s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/450d05c2c3bc4d3eb4276e6705f3e1e0 : SUCCESS in 1h 00m 07s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c67f515b4bc462ba3e1d0366d0140b2 : SUCCESS in 1h 32m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a17a43dfd0244e08d051d61a47ea898 : SUCCESS in 3m 26s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e3c5f039efd44e5eb6e3d1604de369c3 : SUCCESS in 2h 05m 47s","accounts_in_message":[],"_revision_number":3},{"id":"aece0ae96b07b1656adc07d40af445412cbc9ab1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-10 08:03:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a3d1d44a38feb70adc61d7e4854cc3e269488686","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-10 18:49:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4003bbf95f87cad5a5d534641e06aa341c81a918","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 18:51:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a1dba665d877a772d2cad4b98ae37ce938f8c258","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 21:14:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8616776633284ab9a8af234e7ccac6b7 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3908d4034354418d920cf374cef87327 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8287f77629e64af29131d9ff32fe6f66 : SUCCESS in 3m 46s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/147d43568ca64e3c8c884c3e3a312565 : SUCCESS in 11m 09s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a724b6b4de04bf7b0fd56ad0f3b73b3 : SUCCESS in 12m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51e0b620c6034454b73e73f244fc3f52 : SUCCESS in 10m 42s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a78ec8cad3864ac29503e05a18cbb0bc : SUCCESS in 11m 16s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf5c839be2754772960a1eb2fb88bb28 : SUCCESS in 1h 41m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fbfe25941feb4e60808a22b3649dc21b : SUCCESS in 1h 35m 03s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/059398cf7b0945ed8eadd6af33217df3 : SUCCESS in 1h 17m 20s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f184c7c678e486f9e29c7cf235a45ed : SUCCESS in 1h 03m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b9143b0db68482993381c698bc0d1e4 : SUCCESS in 4m 35s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7cc99e8dc3d845a3b5aacd0e83edc1e8 : SUCCESS in 2h 06m 29s","accounts_in_message":[],"_revision_number":3},{"id":"bdfff3a54f3ff2cc55b3c9c3ee0438488c1ff2aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 21:14:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e1e8679fe83edd0ea92d76b1fdd3816225678492","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 21:17:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c2cdb01f2c94087bb3ebbd724cb9d82 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f91cf635606de03cffa657288fe13a7a981b6e20","revisions":{"4b222c16fff94831a5c24bc3ea66be53e76c53f8":{"kind":"REWORK","_number":1,"created":"2022-01-08 03:16:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/823883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/83/823883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/83/823883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/83/823883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/83/823883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/83/823883/1"}}},"commit":{"parents":[{"commit":"87c397708283721602bfe7ef3c5c16110ddbccfe","subject":"Merge \"Accept system scope credentials for Unified Limits API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/87c397708283721602bfe7ef3c5c16110ddbccfe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-08 03:09:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-08 03:15:32.000000000","tz":540},"subject":"barbican_service_user: Accept system scope credential","message":"barbican_service_user: Accept system scope credential\n\nThis change allows usage of system scope credentials in addition to\nproject scope credentials, to use service user token when accessing\nBarbican API.\n\nChange-Id: I3ab30adbae3e805aee36bd32854e5490b19b0097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/4b222c16fff94831a5c24bc3ea66be53e76c53f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/4b222c16fff94831a5c24bc3ea66be53e76c53f8"}]},"branch":"refs/heads/master"},"a474a798c8b68c90d819f3854dc0300212015d10":{"kind":"REWORK","_number":2,"created":"2022-01-08 04:20:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/823883/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/83/823883/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/83/823883/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/83/823883/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/83/823883/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/83/823883/2"}}},"commit":{"parents":[{"commit":"87c397708283721602bfe7ef3c5c16110ddbccfe","subject":"Merge \"Accept system scope credentials for Unified Limits API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/87c397708283721602bfe7ef3c5c16110ddbccfe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-08 03:09:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-08 04:20:06.000000000","tz":540},"subject":"barbican_service_user: Accept system scope credential","message":"barbican_service_user: Accept system scope credential\n\nThis change allows usage of system scope credentials in addition to\nproject scope credentials, to use service user token when accessing\nBarbican API.\n\nChange-Id: I3ab30adbae3e805aee36bd32854e5490b19b0097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/a474a798c8b68c90d819f3854dc0300212015d10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/a474a798c8b68c90d819f3854dc0300212015d10"}]},"branch":"refs/heads/master"},"f91cf635606de03cffa657288fe13a7a981b6e20":{"kind":"REWORK","_number":3,"created":"2022-01-08 06:26:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/823883/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/83/823883/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/83/823883/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/83/823883/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/83/823883/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/83/823883/3"}}},"commit":{"parents":[{"commit":"87c397708283721602bfe7ef3c5c16110ddbccfe","subject":"Merge \"Accept system scope credentials for Unified Limits API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/87c397708283721602bfe7ef3c5c16110ddbccfe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-08 03:09:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-08 06:25:59.000000000","tz":540},"subject":"barbican_service_user: Accept system scope credential","message":"barbican_service_user: Accept system scope credential\n\nThis change allows usage of system scope credentials in addition to\nproject scope credentials, to use service user token when accessing\nBarbican API.\n\nChange-Id: I3ab30adbae3e805aee36bd32854e5490b19b0097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f91cf635606de03cffa657288fe13a7a981b6e20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f91cf635606de03cffa657288fe13a7a981b6e20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
