)]}'
{"id":"openstack%2Fpuppet-oslo~825699","triplet_id":"openstack%2Fpuppet-oslo~master~I0cbd9d22b53ae2f966f0a007231a9003dc03b36e","project":"openstack/puppet-oslo","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-24 18:57:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0cbd9d22b53ae2f966f0a007231a9003dc03b36e","subject":"Ensure kombu_ssl_certfile and kombu_ssl_keyfile are used together","status":"MERGED","created":"2022-01-21 02:22:35.000000000","updated":"2022-01-24 18:57:17.000000000","submitted":"2022-01-24 18:57:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825699","meta_rev_id":"30fede6e170976a24a635fd7db3cd194ced71974","_number":825699,"virtual_id_number":825699,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-24 18:57:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-01-24 08:04:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-01-24 17:13:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-01-24 17:13:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-21 04:13:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-24 08:04:26.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-01-24 17:13:49.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"88e7ac978b0092db0fffe4d7e4f30d9ad1eefd67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 02:22:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f81e170d89bb8d458ec6b5544d5b97aee0b1d14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 02:23:47.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"57c11cb89433a4d1b45a7a8a65555489100e1422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 04:13:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfc4f09e83474536a6f2a99efd7ea763 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/078b8d9517b844dfb702e4cb4e801f2c : SUCCESS in 3m 59s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/630496b4047d45a0b51937fc0f98d3ed : SUCCESS in 3m 48s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c814ea3aa36c4879aa571a5bcc161935 : RETRY_LIMIT in 10m 04s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e82c69ba7c74f1483a860942109d7e2 : RETRY_LIMIT in 11m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/452717ea44964f07951e6dcf193f29af : FAILURE in 11m 40s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae43bee4ebb84accb03ef12b9c786dc6 : FAILURE in 12m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b7f78ecf5ff4ceb9199639fd0e5b85d : FAILURE in 11m 33s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5964cb052fe4565baad3d82d24d02f1 : FAILURE in 37m 43s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44e1fab6ba98439794f3e47630f3897f : FAILURE in 35m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34d9339ccd7d4be1ab81fedb7adc3273 : FAILURE in 12m 17s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20721e97efd24354be8303fae09469d5 : FAILURE in 30m 58s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c81ba5ebdbe44d39888b1dc15131fa48 : RETRY_LIMIT in 7m 43s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8bc9314e110841abba91e6237f97a342 : RETRY_LIMIT in 7m 18s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f842794b09d4807a4162458f658d93b : RETRY_LIMIT in 7m 39s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/868846df5b824bc2b73cae406788fffa : RETRY_LIMIT in 8m 23s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dea674cb7ca04816b9eca0c6a492f4a1 : RETRY_LIMIT in 8m 43s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eedc4d347bab4b49969875cc0297dddb : SUCCESS in 26m 15s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b62be0148ee545f497328fceb92ab392 : SUCCESS in 45m 55s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1b303c7972f4d97971e79429d7929da : SUCCESS in 53m 56s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f078e6b6869a422abfba4d98b38afdab : SUCCESS in 59m 16s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa8e1f8b64304d42954adf171b364b99 : SUCCESS in 1h 04m 03s (non-voting)\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3ca1cd728a6a4cf59e51809d6c0fe717 : RETRY_LIMIT in 6m 37s","accounts_in_message":[],"_revision_number":2},{"id":"f4ba408597c30533d9fd6e56663d93860c6b88b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 05:03:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"375cb9002d985df924d47e48eba64cc56b2d3bdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 05:14:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3094f336168ddcba54588cdab297cb718bc331de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 05:43:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4f142a76dbf4801a359a71063e9791c : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f598db9a080540739b9107c13a7aac4f : SUCCESS in 3m 59s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51a5fe0670374ad78d05b71703506357 : SUCCESS in 4m 16s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f32e745efa0d4944b4d7e2e1b913305c : RETRY_LIMIT in 5m 13s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/af35deb20f6e43699e773e32dab90f54 : RETRY_LIMIT in 4m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc75e93a96c4485a82754e1c43cece16 : FAILURE in 8m 39s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db11b911b5f94c329922f055280bac03 : FAILURE in 7m 41s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7716782c28fd4af3a0c1cbde19291829 : FAILURE in 7m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f3f1c86dd324dbf9bfa3a26588d8835 : FAILURE in 8m 29s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b055f2fa577f4553baf17235d904fc2b : FAILURE in 8m 55s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03eb04a9d6ff4e5aafe5a5b79a5574f9 : FAILURE in 10m 04s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b38b842a81c4b7a956867c33e6ddd0e : FAILURE in 7m 43s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/41f05c3b6e5b44df9e58ecaf999b4a7f : RETRY_LIMIT in 6m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9207843a5314cb391e300f54042298e : RETRY_LIMIT in 5m 07s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5362a6ccf85149cc8e082a5e3d60b837 : RETRY_LIMIT in 8m 09s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/99ab329951ee4c3881f89bcb62663b8e : RETRY_LIMIT in 6m 11s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/34306e0242a14922b9660a00f4359e52 : RETRY_LIMIT in 4m 54s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b00d4e7246be4e19afae4f417b18aabf : FAILURE in 10m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/502dc62f488649ae965fdb5771f5501e : FAILURE in 12m 14s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/474cffc3865343f6abde773c26db6a93 : FAILURE in 12m 02s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ee8cb22d8de4085994ff4d7e5f58031 : FAILURE in 12m 11s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2398a34405a54057b33656f04ce0d13b : FAILURE in 12m 09s (non-voting)\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4f69be03d021412eb3af9c15b68d6920 : RETRY_LIMIT in 6m 08s","accounts_in_message":[],"_revision_number":4},{"id":"404a90e3ee19d3f04543c875a018264bbeae6c9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 06:06:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a07b209e2c4ce271828bcb3ed3454968ee8925e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 07:20:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56ad35c7dec149beafe20a07e30282b7 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6164f78a2605443d992517251c25c735 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f9b70816cbc4890b6182659f25b6b15 : SUCCESS in 3m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b22baacf6f84b1aa02f03ecc41f4fd1 : RETRY_LIMIT in 5m 57s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7fb20d3e5ccb414ca4a16bb5ccee2b09 : RETRY_LIMIT in 4m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9302a47c4ff94c9288b7c65dff523ce5 : SUCCESS in 7m 19s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11a3d7abcbf64005999a4b8bad080cfd : SUCCESS in 7m 57s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99f0a12ad81546219f614ecfe5bc8c58 : FAILURE in 9m 48s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dab31e1dcb2b49e9a136db9770d4f7ca : FAILURE in 30m 40s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f0c33eca5514858a6008024322860dc : FAILURE in 32m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1996cd0088a84940bf86c20620d6122a : FAILURE in 6m 17s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bf6077bbadb4613afcc99d93022aaac : FAILURE in 27m 40s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f1002fe0a0fa4af4a7cba6935d8e9652 : RETRY_LIMIT in 5m 29s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dae996f374dd4db6aca43723b1209d5b : RETRY_LIMIT in 7m 03s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2033bc85d9d44dd8b2a4bcad17941d5c : RETRY_LIMIT in 5m 21s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/75ef065f3f8b40268dc493ad097f8712 : RETRY_LIMIT in 5m 27s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/daaa293e1d8b4d1f8430e8fb24977b7d : RETRY_LIMIT in 5m 49s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7fae253b0ce74318b8640293fddecc8b : SUCCESS in 21m 04s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7f06907e44946e3a3648e22fd6f971b : SUCCESS in 45m 53s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4ad798042d1448c83837839f3e2119f : SUCCESS in 43m 31s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e65c207209914652bf632e9c916c867c : SUCCESS in 53m 23s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ffb732ff6cf4866ab9945371733a7b7 : SUCCESS in 42m 16s (non-voting)\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9e768ef70569451d9a5dcd3625ec9f44 : RETRY_LIMIT in 4m 32s","accounts_in_message":[],"_revision_number":5},{"id":"a5853c7fbaad087c1c651de92e7aac3275156f71","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-22 12:10:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d743fad246a9ac0f3f44e00d702c2721ee5765d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-22 14:13:26.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54e5f2b41dae4a1eb199d33366b865ff : SUCCESS in 3m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ba87a0b194740c896884b12857444ab : SUCCESS in 6m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8917555067c4156a43bcd318ce6a44d : SUCCESS in 3m 19s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b550b9ab7a84c7bbc48a1e401d9b83a : SUCCESS in 8m 53s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca80a543024540c8a8192e92251b2928 : SUCCESS in 10m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/def55613c5bd45e48599eaa57cf675b8 : SUCCESS in 11m 31s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b81ebd8d15af4de9962e07097f3c82f0 : SUCCESS in 15m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a855d7c54bd443cb1d10ef0e92699e2 : SUCCESS in 7m 09s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/876e0958cb684c9f8deb1c9a670e93dc : FAILURE in 30m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d20d7cb947284a5db937013ee312c95b : FAILURE in 40m 25s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/319550a39b644e5f852776242a372ee5 : FAILURE in 7m 01s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/556376e0cc724efdb9a9eac6ba2a838a : FAILURE in 20m 40s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e10a4384f4854e3ea27ba87f289f8548 : SUCCESS in 21m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/99e157e167e148e48e7b658ea21991ea : SUCCESS in 1h 25m 30s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d97f16d2249445fb392affb486534fc : POST_FAILURE in 1h 30m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2039ea8948574d6fa3388e06d7a5393d : SUCCESS in 1h 42m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dda7a47a62864c46b7f38f485387e7ea : SUCCESS in 1h 12m 27s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5533ec4f16b14499bc9ce80900d4d6fb : SUCCESS in 17m 51s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac2eb98655db453e92054eea812b9942 : SUCCESS in 53m 53s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0d454bd320c4b80aca34a04836733a0 : SUCCESS in 1h 01m 03s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7016c0022e54aab80e87a7aa717c272 : SUCCESS in 45m 39s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/be9485ac101c417e9ab788469e97feb5 : SUCCESS in 40m 51s (non-voting)\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/457ceaeddecf46e7b85cd89152207c1d : SUCCESS in 2h 01m 27s","accounts_in_message":[],"_revision_number":5},{"id":"176b66cff0d55233efcbf9ecef132585916c45bc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-23 09:27:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"38475f47665d152b6afd0dd0e70d96404fd05d14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-23 11:50:20.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faa6000c856d4168908b7430640b0ab0 : SUCCESS in 5m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ade3e068b0e48a0945692b76437b3fb : FAILURE in 16m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d35dd0bb64f849d8967f343a272f3fe9 : SUCCESS in 3m 43s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1efa0ef65e0346dcaea0c14046c8dfe5 : SUCCESS in 9m 38s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd0f76d35b574e4b91b660b9c62c340e : SUCCESS in 9m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3a75c6f49e1422294afda522b104a56 : SUCCESS in 7m 13s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92fa96996f2945dcbbdc7c53a77a945e : SUCCESS in 6m 51s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe61fcc6e57e40a8a3625d4d606dc206 : SUCCESS in 6m 47s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdb4c725265b453d982363a1001e76a0 : FAILURE in 34m 38s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb77b0751b8248af8941916f929eabbf : FAILURE in 36m 28s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1baad4c0075c4230a43e382d0cf1a32f : FAILURE in 6m 49s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c40c8d0d77f646a6a87c849112019818 : FAILURE in 25m 39s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/20e65dee3325405b862f1effe64f7006 : SUCCESS in 23m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c287ade7682244da87a34e87141adb5d : SUCCESS in 1h 31m 34s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ff3fc6d89654835bf49fde4af716035 : TIMED_OUT in 2h 02m 58s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a26e1a4d54a43ad877d9fa1cba93799 : SUCCESS in 1h 26m 26s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a48fe14ec8947a9ac651ed07ff2412e : SUCCESS in 1h 00m 32s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d64453768db49df8cc290e33e27f219 : SUCCESS in 18m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/57ab1a9b9d274bb98061fec614ac2ae1 : SUCCESS in 49m 38s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe8ec2aef4a645ed8340bb69d0c7a1d3 : SUCCESS in 1h 06m 01s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/07404e6099a943b1aefe505e5cff265c : SUCCESS in 1h 04m 39s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/43b8bbb10350474dae80d16b98d9c215 : SUCCESS in 42m 13s (non-voting)\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ddb108ae47564c02913dcab21e5940aa : SUCCESS in 2h 20m 27s","accounts_in_message":[],"_revision_number":5},{"id":"5d1f1b7534d423167be4b29e66318d5462a49141","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-24 00:59:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"389f74ee15b1536e03eabf7e047a1e3e4751c51f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 03:06:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cbf612372904890918842053eb047f4 : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c088f2b7b0245709b3f35b455198281 : SUCCESS in 3m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32875068958641caa0a463c8f2948cfc : SUCCESS in 3m 24s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/621eb70ebf9340ff85b24fbdf46e245d : SUCCESS in 9m 40s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b9b16f6c6034b54b83cca6e6c8d08b6 : SUCCESS in 9m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffadf199fff6436b8b57aded1f1ca9c7 : SUCCESS in 7m 14s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed04b1d12c6d4c299d676c88e8ced02c : SUCCESS in 7m 45s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66e7868208cf4b0f803e3e22ca27270a : SUCCESS in 6m 55s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0497b0adfe374ef9af6b613f98232ac8 : FAILURE in 30m 18s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bcda5d667604ebbac4b45d5d6791780 : FAILURE in 38m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cead37dbfa8a4369a1be0818df5293f9 : FAILURE in 6m 52s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d90771e5ae234f56a1432ff07e4a27ff : FAILURE in 24m 07s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/68a10d904566419f8a62e03055a43312 : SUCCESS in 21m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6076b7244f1048209dfb999b672b1a9f : SUCCESS in 1h 16m 56s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f667fe6f403f45fb99002881ca7fc4fa : SUCCESS in 1h 30m 39s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/19330adffcc546d08d16876f2c45b9ef : SUCCESS in 1h 23m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c506b3ed59ad485ea571382184f64187 : SUCCESS in 1h 03m 31s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b9f4643ef7f4791827b3fc07f410645 : SUCCESS in 17m 48s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/254823889e4a41119f82a63614c6ad20 : SUCCESS in 51m 41s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/335a44ed4a5a49b78887dd37eaa05984 : SUCCESS in 55m 09s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/efad401368ad46d4b6b3e4cc7760269a : SUCCESS in 57m 48s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2cad25f71564ed79a08d62ee6d6817f : SUCCESS in 42m 42s (non-voting)\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6b0afed5ad1d47eba1791e03e76251d2 : SUCCESS in 2h 01m 41s","accounts_in_message":[],"_revision_number":5},{"id":"92be234e3373f7f60529ce736256782ebf759d06","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-24 08:04:26.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6b1eb0918c48bc44c6e979bb9b84d0339da22aed","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-24 17:13:49.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8ba7faca9c58e480830df84969629fa338f92621","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 17:15:36.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"84eec979f206e70932672e02429d75b77995b77f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 18:57:17.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a480bcb3c644fb0b4bf2e32fc5eaf1b : SUCCESS in 4m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b094f06a1e5a4ec098f61b97b3c99da0 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f251c2ffad941c8946785c009e88af4 : SUCCESS in 5m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc6c75621001415e93b89754cba4ab40 : SUCCESS in 13m 27s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9bc47b70270548469c188240aa0e2eb2 : SUCCESS in 13m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9970b41ab8a433683c4b8774fdf6013 : SUCCESS in 9m 06s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e56b5370210f4d9b9dcc9565985ea142 : SUCCESS in 9m 01s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/edafb85422dc4b5db6f1b94674396da3 : SUCCESS in 1h 09m 14s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/55dee7e68ea442cf97060b96f28ef76b : SUCCESS in 1h 29m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb6de38500a047dfaa30c4deecc45b03 : SUCCESS in 1h 12m 05s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/79e6732733f34dfebfe506f89c2a33e9 : SUCCESS in 1h 06m 47s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e73cc3e008ca47c4a294a641c8e9f754 : SUCCESS in 1h 38m 52s","accounts_in_message":[],"_revision_number":5},{"id":"30fede6e170976a24a635fd7db3cd194ced71974","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 18:57:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f898c0f8ddbd77601c9f2c49050bc478cc081c35","revisions":{"b1486ec172e541e186c8282d1350027223cf0459":{"kind":"REWORK","_number":1,"created":"2022-01-21 02:22:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/825699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/99/825699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/1"}}},"commit":{"parents":[{"commit":"f91cf635606de03cffa657288fe13a7a981b6e20","subject":"barbican_service_user: Accept system scope credential","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f91cf635606de03cffa657288fe13a7a981b6e20"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 02:22:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 02:22:11.000000000","tz":540},"subject":"Ensure kombu_ssl_certfile and kombu_ssl_keyfile are used togather","message":"Ensure kombu_ssl_certfile and kombu_ssl_keyfile are used togather\n\nChange-Id: I0cbd9d22b53ae2f966f0a007231a9003dc03b36e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b1486ec172e541e186c8282d1350027223cf0459"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b1486ec172e541e186c8282d1350027223cf0459"}]},"branch":"refs/heads/master"},"c4b9e644343d839276c02a50c71ca09c883dd623":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-01-21 02:23:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/825699/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/99/825699/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/2"}}},"commit":{"parents":[{"commit":"f91cf635606de03cffa657288fe13a7a981b6e20","subject":"barbican_service_user: Accept system scope credential","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f91cf635606de03cffa657288fe13a7a981b6e20"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 02:22:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 02:22:56.000000000","tz":540},"subject":"Ensure kombu_ssl_certfile and kombu_ssl_keyfile are used together","message":"Ensure kombu_ssl_certfile and kombu_ssl_keyfile are used together\n\nThis change migrate the validation from puppet-designate/neutron to\nthe common place, because the requirement is valid for all components.\n\nChange-Id: I0cbd9d22b53ae2f966f0a007231a9003dc03b36e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c4b9e644343d839276c02a50c71ca09c883dd623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c4b9e644343d839276c02a50c71ca09c883dd623"}]},"branch":"refs/heads/master"},"3a0571671c154d04d6ec9194baa29a01ed20c956":{"kind":"REWORK","_number":3,"created":"2022-01-21 05:03:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/825699/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/99/825699/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/3"}}},"commit":{"parents":[{"commit":"f91cf635606de03cffa657288fe13a7a981b6e20","subject":"barbican_service_user: Accept system scope credential","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f91cf635606de03cffa657288fe13a7a981b6e20"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 02:22:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 05:02:56.000000000","tz":540},"subject":"Ensure kombu_ssl_certfile and kombu_ssl_keyfile are used together","message":"Ensure kombu_ssl_certfile and kombu_ssl_keyfile are used together\n\nThis change migrate the validation from puppet-designate/neutron to\nthe common place, because the requirement is valid for all components.\n\nChange-Id: I0cbd9d22b53ae2f966f0a007231a9003dc03b36e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3a0571671c154d04d6ec9194baa29a01ed20c956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3a0571671c154d04d6ec9194baa29a01ed20c956"}]},"branch":"refs/heads/master"},"48ee291688c8389acaeb9a206c626c0e8796b6e9":{"kind":"REWORK","_number":4,"created":"2022-01-21 05:14:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/825699/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/99/825699/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/4"}}},"commit":{"parents":[{"commit":"f91cf635606de03cffa657288fe13a7a981b6e20","subject":"barbican_service_user: Accept system scope credential","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f91cf635606de03cffa657288fe13a7a981b6e20"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 02:22:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 05:14:43.000000000","tz":540},"subject":"Ensure kombu_ssl_certfile and kombu_ssl_keyfile are used together","message":"Ensure kombu_ssl_certfile and kombu_ssl_keyfile are used together\n\nThis change migrate the validation from puppet-designate/neutron to\nthe common place, because the requirement is valid for all components.\n\nChange-Id: I0cbd9d22b53ae2f966f0a007231a9003dc03b36e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/48ee291688c8389acaeb9a206c626c0e8796b6e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/48ee291688c8389acaeb9a206c626c0e8796b6e9"}]},"branch":"refs/heads/master"},"f898c0f8ddbd77601c9f2c49050bc478cc081c35":{"kind":"REWORK","_number":5,"created":"2022-01-21 06:06:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/825699/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/99/825699/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/99/825699/5"}}},"commit":{"parents":[{"commit":"f91cf635606de03cffa657288fe13a7a981b6e20","subject":"barbican_service_user: Accept system scope credential","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f91cf635606de03cffa657288fe13a7a981b6e20"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 02:22:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-21 06:06:25.000000000","tz":540},"subject":"Ensure kombu_ssl_certfile and kombu_ssl_keyfile are used together","message":"Ensure kombu_ssl_certfile and kombu_ssl_keyfile are used together\n\nThis change migrate the validation from puppet-designate/neutron to\nthe common place, because the requirement is valid for all components.\n\nChange-Id: I0cbd9d22b53ae2f966f0a007231a9003dc03b36e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f898c0f8ddbd77601c9f2c49050bc478cc081c35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f898c0f8ddbd77601c9f2c49050bc478cc081c35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
