)]}'
{"id":"openstack%2Fpuppet-oslo~826869","triplet_id":"openstack%2Fpuppet-oslo~master~If4622f85e1e7f715805b205bfc3c16cb1e53b943","project":"openstack/puppet-oslo","branch":"master","topic":"oslo.cache.2.10.0","attention_set":{},"removed_from_attention_set":{"31863":{"account":{"_account_id":31863,"name":"Grzegorz Grasza","email":"grzegorz@grasza.com"},"last_update":"2022-01-29 00:17:44.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-28 22:18:14.000000000","reason":"Takashi Kajinami replied on the change"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-01-29 00:17:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4622f85e1e7f715805b205bfc3c16cb1e53b943","subject":"Add retrying options to oslo.cache","status":"MERGED","created":"2022-01-28 10:26:25.000000000","updated":"2022-01-29 00:19:03.000000000","submitted":"2022-01-29 00:17:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826869-oslo.cache.2.10.0","meta_rev_id":"625d61d02f292047bf21e6baadc268fbaf888d46","_number":826869,"virtual_id_number":826869,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-29 00:17:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-01-28 22:18:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-01-28 20:47:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-01-28 22:18:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31863,"name":"Grzegorz Grasza","email":"grzegorz@grasza.com"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31863,"name":"Grzegorz Grasza","email":"grzegorz@grasza.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-28 10:27:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":31863,"name":"Grzegorz Grasza","email":"grzegorz@grasza.com"},"state":"REVIEWER"},{"updated":"2022-01-28 10:27:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-01-28 12:45:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-28 20:47:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"56eae5a016c37fc6088204699808e198140cdb4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-28 10:26:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fe6c6f28231b8fdfabaa6e1b576af76c19a6556","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 12:45:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/602cdbf8e02444cda559fb372ee27851 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18ca3bf58d1044a08ebd5c1cee1b37a9 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf8edbb85b504c1884d6c071de221eee : SUCCESS in 3m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1f97f041d1e4662a6a98fa4209edef7 : SUCCESS in 11m 53s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc0cc418a6414373acf4a5f572a3b8f4 : SUCCESS in 11m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5dc355b6540487985522d011a130e13 : SUCCESS in 8m 17s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89ab4ec4ada544aa95f401445788e27e : SUCCESS in 8m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dab598a1a0fe44a6a9af93257b3c7cda : SUCCESS in 8m 29s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5af651d0ad64676b30cf6fe3de2c6d5 : FAILURE in 31m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fac595ff8a3c4d2ab0e7d1a458920293 : FAILURE in 39m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/772fdbdacda84097ad75f3794ce10d7d : FAILURE in 35m 12s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/549be1835321400095e3d1547e63aea2 : FAILURE in 38m 20s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5cfe17d7f86c475e8f70b6b675588656 : SUCCESS in 21m 42s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9c9c30483d34215bbace41390bb3e85 : SUCCESS in 1h 09m 48s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/44d2e52e7b8e470286e1812ed62acc90 : SUCCESS in 1h 35m 28s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b6a57219f3c44f7996d5a7adb04f822c : SUCCESS in 1h 14m 05s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/604357dc8ac343f4a1ab0ad8d575a900 : SUCCESS in 59m 35s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/62e7e606ff9c4b80ae5409ffa0f1e615 : SUCCESS in 19m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e08144034d34473a983e23380525520 : FAILURE in 52m 46s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/95fdfe8f856a4ec3bad50fe28f985384 : FAILURE in 50m 01s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/53cc2e4078294c47a6bea8174fe23464 : FAILURE in 54m 05s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8f7a4748fcf473db8ed06e441df1a34 : FAILURE in 44m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b85df5fe52140b4ba684a27cc517a1c : SUCCESS in 3m 06s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fce03ffef4084cdb9bd35b13fcfc3f80 : SUCCESS in 2h 16m 09s","accounts_in_message":[],"_revision_number":1},{"id":"522c214d518327df23b0631f722d66e4cf40fa5c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-28 20:47:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bfea1527a63b81da4d9a72ebeae3bd21e385e6c7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-28 22:18:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7eb27345e2b42b53c382614ebe06e792805fbc9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 22:20:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3acb497bd74a412c6e9a470b1ec36e49b9b60d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 00:17:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62231387f96c42fcb3f415aa157dda74 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e0826480ed04f1bac600d2e9225ee96 : SUCCESS in 3m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/463ee5120790423d9d8ba343ff25e806 : SUCCESS in 3m 46s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/630cab50c357471ebe4ab0a3aac55126 : SUCCESS in 9m 56s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/0825a56455a04100bed32c055b995527 : SUCCESS in 11m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/229acea8a1654468872df80939d4c351 : SUCCESS in 7m 42s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa3616cef7d84f2eb8a12df6c0fb1dbf : SUCCESS in 8m 05s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7b273351f7149a58d8c370c76e66fde : SUCCESS in 1h 10m 35s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/35138d818b7d4d329bbfa5db72d22b0e : SUCCESS in 1h 19m 20s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8c4215c9cb944b22b75a28479481c617 : SUCCESS in 1h 13m 28s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/37e23dc32f6646258c19d96d4b90e289 : SUCCESS in 1h 00m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dde80c350e544039933c80cb32ad30d9 : SUCCESS in 3m 31s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8535fdc46f1f492382533af297c51025 : SUCCESS in 1h 56m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d9dcd77da5264b1024380995d2051acd33771329","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 00:17:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"625d61d02f292047bf21e6baadc268fbaf888d46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 00:19:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f48e6b8b8c2c4edf94a572c93c545f97 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7dbe8aa5f4ebc31d77344932fef2eea84a32906","revisions":{"f7dbe8aa5f4ebc31d77344932fef2eea84a32906":{"kind":"REWORK","_number":1,"created":"2022-01-28 10:26:25.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/69/826869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/69/826869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/826869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/826869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/826869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/69/826869/1"}}},"commit":{"parents":[{"commit":"19ad5b2fa1968846c54ef46e8182aa078f7c9226","subject":"Merge \"Add socket keepalive options for the pymemcache backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/19ad5b2fa1968846c54ef46e8182aa078f7c9226"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-01-28 10:24:04.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-01-28 10:24:04.000000000","tz":60},"subject":"Add retrying options to oslo.cache","message":"Add retrying options to oslo.cache\n\nThis patch specifies a set of options required to setup the retrying wrapper\nfeature of pymemcache (dogpile.cache) cache backend.\n\nOriginal oslo.cache change:\nhttps://review.opendev.org/c/openstack/oslo.cache/+/803747\n\nChange-Id: If4622f85e1e7f715805b205bfc3c16cb1e53b943\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f7dbe8aa5f4ebc31d77344932fef2eea84a32906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f7dbe8aa5f4ebc31d77344932fef2eea84a32906"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
