)]}'
{"id":"openstack%2Fpuppet-oslo~827035","triplet_id":"openstack%2Fpuppet-oslo~stable%2Fxena~I90d8e2b0cf231c9bd0fe97d4f363795886ccf8c7","project":"openstack/puppet-oslo","branch":"stable/xena","topic":"xena","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-02-04 13:42:33.000000000","reason":"Hervé Beraud replied on the change"}},"hashtags":[],"change_id":"I90d8e2b0cf231c9bd0fe97d4f363795886ccf8c7","subject":"Add pymemcache\u0027s HashClient retrying options to oslo.cache","status":"MERGED","created":"2022-01-31 10:12:22.000000000","updated":"2022-02-04 16:46:13.000000000","submitted":"2022-02-04 16:43:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827035-xena","meta_rev_id":"929e8b075fa14d6b219962ef2d479794842176d0","_number":827035,"virtual_id_number":827035,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-04 16:43:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-02-03 15:31:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-02-04 08:21:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-02-04 08:21:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"date":"2022-02-04 13:42:33.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-31 11:43:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-03 15:31:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-02-04 08:21:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"7656cbb1ed2ff699a2e50cabfbcc150519c16ca3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-31 10:12:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b247eb22ddca6b8c7183692d9a28c0c2edf47db4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-31 10:13:06.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cdf15a5424d2a1cae1a05b272ee920731d4146b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 11:43:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce3fe573821a4a41a561e3a6cd4d1750 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8714d9c3ff524debb9d65bb060415069 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c45e87bec4994beeaf520a5b1e096800 : SUCCESS in 3m 28s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5645a84398fb4439841314991f8bafa0 : SUCCESS in 10m 35s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1a84d3d02e74477a132d807150d6663 : SUCCESS in 9m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d92e7b746e04b03b570ae1589029890 : SUCCESS in 7m 23s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b595913039834a8bb85dd05bd0ae2f0a : SUCCESS in 7m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f64901ce4cdb48568ef3665b83c7ebfb : SUCCESS in 7m 01s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a65a5b9b11954c249115fad3fe007d3b : FAILURE in 37m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/493e45cf0f2a469c81fd073aa45fb150 : FAILURE in 36m 13s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/467bdb8cb3ad45a38234a48858b1c558 : FAILURE in 33m 07s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a77e1c61bcc24939a446516508dac566 : FAILURE in 26m 48s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e9e645769c44fb29b0347fb7d0858fc : SUCCESS in 21m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e82e614eca95423fad5d9b573749201a : SUCCESS in 1h 14m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d4837f1529d4a28b3e728fa41a40908 : SUCCESS in 1h 25m 13s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/35e4339d6107476f968083824e8a7622 : SUCCESS in 1h 17m 03s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1dd57b2e50c4475a0d06f24d8fcbc6d : SUCCESS in 59m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0a2a894b24247c58393e39c4cfa5164 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b1e3696cab5bb10383bd037171549bf668f28d91","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-03 07:21:59.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9fd38c8397bd3b612191049cf6e0064e5cd50998","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-03 15:31:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d2dd4b2937ad6c2e86badb31c8320f1510426674","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-04 08:21:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a1b89357551a0a1bdfd7a09e9bd0f5e1108f5dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 08:23:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cea4d78096618909188f9a8d5fe96cb2ea92164a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 10:28:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/032efb1bc5cc4e09b4faf83604d58f0b : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dc09ff0757f40bb870c625ab04820cd : SUCCESS in 4m 41s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37d1046e34e54eeabe755a392c9c2ec2 : SUCCESS in 4m 15s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/47cd2ea59f3b42fea6cd4e7e4238306c : SUCCESS in 13m 17s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/29a5f435485b440fb72384c0f4e7add0 : SUCCESS in 10m 47s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99ef3705b47b438e9554f01b23a027fa : SUCCESS in 11m 13s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d45d28115644b9696e286ccf549b77a : SUCCESS in 8m 19s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c4e38c42b884528980235aa7fd1c9bf : SUCCESS in 1h 26m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8718c1fa8ce4cb389b51ac4bf30f6f1 : FAILURE in 1h 30m 12s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d294998f74b47e383d5b869b8856e49 : SUCCESS in 1h 30m 58s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/70019e8d9fbe49e28297afcfb58d5e36 : SUCCESS in 55m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/164ebca2c9d148479942ffdd5e48ddb8 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1bd8725ea9dbd27a55b1d8ea4a4bf8f0d5dc8d44","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-04 13:42:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7ae8124f6eefae50f12b68a8ee6ad60463fc88b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 15:11:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7273efe7bd347a3a490b3351a82b91f : SUCCESS in 5m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a70b90709cc644d9bbee82452e04dcbd : SUCCESS in 3m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec41ae76e45d463ab8b1849b7116f2b5 : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f07d4f27d9184ad7ad2a79bfb122ef86 : SUCCESS in 11m 11s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c88ecda9af6a4324a18cf0b43d8ba76a : SUCCESS in 11m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cb023e932ea4eb183bbea88322c31b6 : SUCCESS in 9m 17s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07b66912e9f249f5a35633c7c09c7809 : SUCCESS in 10m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b49a5d284b954715903c5e320c7a6649 : SUCCESS in 9m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/257b1cfe03f240d191ec5e03e8279bab : FAILURE in 43m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9a0d6fd310c434d9dd2573232097e4f : FAILURE in 40m 09s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a949e79d9b1644978eea9df26f6974fe : FAILURE in 33m 44s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8064973bc0de4ddba947184b4ca22abb : FAILURE in 25m 58s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4553aa79678d452da47e4e53320d41e9 : SUCCESS in 22m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4906269b9964e09af0cb04c917d17ef : SUCCESS in 1h 12m 06s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f14b7b801c84243863c144703b62a06 : SUCCESS in 1h 22m 21s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f0932cc46274ba7bb0d174cf62306b4 : SUCCESS in 1h 16m 52s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d048c0eee79b4cd5a1f3e78c8f8c6efc : SUCCESS in 1h 06m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1939ccb815de467b9abe8b88d917c440 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a6ffdea65ae3dee9fccd3f88a65f1e281c03e4ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 15:13:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d723bf04a704f08ffbf6584d82ad3279b8d2447e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 16:43:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f80bdf57a00e4aec8e4066fd42ceaf37 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94b406968a2e49bba8d6f129b36ea8b6 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae12e039b2ab47d2b429fca35602c861 : SUCCESS in 3m 25s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/51ac1c2991ab488fae726b4eebfc6fa6 : SUCCESS in 11m 19s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5c69275b2d5b4187b71f660256caee80 : SUCCESS in 11m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a55783338a640eca1d5bca9951728e2 : SUCCESS in 8m 03s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0d86565748f49cea7a9668fa13681bb : SUCCESS in 8m 50s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/75a94997fc994782b277dc3c9003a930 : SUCCESS in 1h 13m 51s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/903af71741004dc4a0626f51116cb61e : SUCCESS in 1h 18m 56s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/847f89f7d9bc4b9691bfbd0785bd0f4c : SUCCESS in 1h 21m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/97039a4c5348464c8d7ad37ec4a4d294 : SUCCESS in 1h 02m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/323bfdc86b1a407cbdf74dfd304a210a : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"bc205ca075968dd79f1dd1c38635f828daa8b3c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 16:43:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"929e8b075fa14d6b219962ef2d479794842176d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 16:46:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5f25d2aeb4c4446927d55181a7f860b : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"edf16447f0a8fa8be0000a3ecf7d751142048a49","revisions":{"edf16447f0a8fa8be0000a3ecf7d751142048a49":{"kind":"REWORK","_number":1,"created":"2022-01-31 10:12:22.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/35/827035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/35/827035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/35/827035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/35/827035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/35/827035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/35/827035/1"}}},"commit":{"parents":[{"commit":"88197f306123c521967b5a16b4909e5227554325","subject":"Add retrying options to oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/88197f306123c521967b5a16b4909e5227554325"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-01-28 10:39:42.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-01-31 10:02:23.000000000","tz":60},"subject":"Add pymemcache\u0027s HashClient retrying options to oslo.cache","message":"Add pymemcache\u0027s HashClient retrying options to oslo.cache\n\nThis patch specifies a set of options required to setup pymemcache\u0027s\nHashClient retrying feature (dogpile.cache) cache backend.\n\nOriginal oslo.cache change:\nhttps://review.opendev.org/c/openstack/oslo.cache/+/824944\n\nChange-Id: I90d8e2b0cf231c9bd0fe97d4f363795886ccf8c7\n(cherry picked from commit 34e1446f44d66a96c8fbe631fc29413de0a51551)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/edf16447f0a8fa8be0000a3ecf7d751142048a49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/edf16447f0a8fa8be0000a3ecf7d751142048a49"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
