)]}'
{"id":"openstack%2Fpuppet-oslo~827947","triplet_id":"openstack%2Fpuppet-oslo~master~Icaf88ebaaf72d6810d9ded119a9998538eb09869","project":"openstack/puppet-oslo","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-20 11:43:38.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-02-07 12:48:24.000000000","reason":"Hervé Beraud replied on the change"}},"hashtags":[],"change_id":"Icaf88ebaaf72d6810d9ded119a9998538eb09869","subject":"privsep: Allow customizing section name","status":"MERGED","created":"2022-02-05 10:48:50.000000000","updated":"2022-02-20 11:45:41.000000000","submitted":"2022-02-20 11:43:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827947","meta_rev_id":"fc824ffcd7689d1f22a35363f00937b2b2cccf4d","_number":827947,"virtual_id_number":827947,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-20 11:43:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-02-07 15:14:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-02-20 10:01:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-02-07 12:48:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-02-20 10:01:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-05 13:01:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-07 10:30:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-07 15:14:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-02-20 10:01:15.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"b98618f22281f32f188c6f6bb7464211f3bfae20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-05 10:48:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7bc1ee04c968eafe21b7f9faf32b1e10d751088","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-05 10:53:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9015dd1daec7213517102b63ead621e5b247c707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 13:01:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fd835f101214c4e8b733685bfde35fd : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9e576e6ee684112bf07d8eb96de0138 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32d120b20d594590905f65a8f1da56dd : SUCCESS in 3m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0be1e7c6c56e48d48720e46a68504ae6 : SUCCESS in 10m 40s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8887aac968d4b00849192d12b7b45f5 : SUCCESS in 10m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3139af368cec4dc8b24141e2025d0f8d : SUCCESS in 8m 49s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9235f35870e64cbdaa14ae0b644d9004 : SUCCESS in 7m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dbb9231021a436b90ed0afa1852b66a : SUCCESS in 8m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8c30f1df553401c954877310b668b67 : FAILURE in 31m 13s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85069e5f05b246619c78e26c37157144 : FAILURE in 36m 56s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fecb7885cef84eb580e1689304d527cf : FAILURE in 30m 59s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8519a811fa0427bbb82a6b1b85aabd8 : FAILURE in 27m 16s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3779fbc77f334d51838402ca6796b514 : SUCCESS in 20m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a8d8569e1c44290ac8254fd6b81e77d : SUCCESS in 1h 19m 22s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/36a8d6846f4e4944b15f7a1a0e768eb2 : SUCCESS in 1h 17m 04s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/90d3cf13f3c643b0a0d6807bc74b1228 : SUCCESS in 1h 26m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f73294c84b8344108961b388b602ed21 : SUCCESS in 56m 45s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5982b4bab2674acba7447a2168a9bfdb : SUCCESS in 19m 36s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9be2ce3707c40b3949ed0f912187468 : SUCCESS in 57m 29s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec1b66bd8f834081a0fb7f87ebc0ed49 : SUCCESS in 46m 27s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b283b767b91e442fbdf4cb64fbe856d7 : SUCCESS in 42m 49s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b65c8fcd2b742f69d48530cf7fd6e80 : SUCCESS in 32m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a75f9a8611e14d2fa64901b2c6891164 : SUCCESS in 3m 32s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ae528b852ced4e3b94e2e90fa87feb6f : SUCCESS in 2h 06m 08s","accounts_in_message":[],"_revision_number":2},{"id":"03f4425b6bdeeab6a43881fc504d579cb7b276ac","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-07 10:30:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6de1ea4faf7cdbf7708a36b1f17187ae1fdaf5e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-07 10:42:33.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"9af1ef69985dd34f5f104c86aad0a0c2435fd3e0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-07 10:42:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b14d05e2813f8700cbf343540296b80d8101a371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 12:48:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/179861644a8540b6a6ad340bd2d9a592 : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fee566592011469fb838c15a53f30bd0 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d208532aeddf4916a1a9e7e1a42a8f63 : SUCCESS in 3m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba63fa39671c4969811a6a3cf5bbd32d : SUCCESS in 12m 32s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/079504c1545840d89e535c75b14d9e41 : SUCCESS in 9m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c86dcdb031894d2b9d9f2358458d07f4 : SUCCESS in 9m 58s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b32e7d39add9469d9a21440a02f85442 : SUCCESS in 7m 22s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb0c39548d83452490d78a71108dda70 : SUCCESS in 10m 31s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8a482c4b93945a0956bcca44333ee18 : FAILURE in 32m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0937df33deca4dfdb13fd5fa5b5a6bff : FAILURE in 44m 42s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3eac7060af834ff89978a66d66d1858b : FAILURE in 24m 13s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a436637792064e18ac8437f59088fb41 : FAILURE in 31m 17s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f568f1580dc491ca9c0804956b95392 : SUCCESS in 24m 32s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b5af692fc3245ef88d56c0ba4d44cea : SUCCESS in 1h 13m 02s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/77638edf9603437c99e291e7a9b1625d : SUCCESS in 1h 07m 38s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ebb991bba8c647daba1a25de8d42061c : SUCCESS in 1h 25m 54s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6fb568ff7f94f8096f01372cdc7c7cf : SUCCESS in 1h 07m 57s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0c5ff4605c44b688ab29efbfee01e6d : SUCCESS in 19m 36s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c424ac517f444669b40d207bbf2e4b7e : SUCCESS in 57m 23s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d732dd9d1134beca95306830dc5230a : SUCCESS in 1h 00m 07s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7569a9eea4d46289a37f9421186b856 : SUCCESS in 51m 58s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d21def0e5f6d4fa7a13f0e811aff3bcb : SUCCESS in 48m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e75c67d76c39468ead2d2e30a44f71d5 : SUCCESS in 3m 56s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2a17677ad78e43adb18ed7d7e92db3fb : SUCCESS in 2h 00m 51s","accounts_in_message":[],"_revision_number":3},{"id":"b8f05e1f286b00074f17a4b629639835a16b2206","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-07 12:48:24.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"64d9187c69d96296cba56791190039edc22038ad","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-07 15:14:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3eb080bd077a408145dbb2d1c8217ce566682736","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-20 10:01:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"405cbf248cab8b1911cd79935166f6c28b3a7b8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 10:02:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"12c8adb7d11aa4b28a2ba3c6f5b35f9484c2ad28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 11:43:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c079d3293e542038f9a67c92fcb63d1 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66f91a28f1584ee0a1a15ba3a618a206 : SUCCESS in 3m 07s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2801b0d195eb42998e6832a378e503c8 : SUCCESS in 3m 39s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/63fd891d1fa9425ca785f50e8e36282f : SUCCESS in 9m 55s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/26d950c079d64ef885cfbf3a1b91987c : SUCCESS in 10m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0bfe988c0054367a5214adddc348596 : SUCCESS in 10m 00s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/626bae10b9634f03bfd74196ee0ed367 : SUCCESS in 8m 12s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f05a78512734a16a8dc382cb1c0766a : SUCCESS in 1h 17m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/324588706e0f4d77b445d0ee8deb6d8f : SUCCESS in 1h 39m 47s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f6e9003bee5749c99f063de6dae7bbc1 : SUCCESS in 1h 33m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c6c31003b6a456ea5d95d20b985ed4b : SUCCESS in 1h 02m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a07fea5c506e43fba04de34ae39cf164 : SUCCESS in 3m 47s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/73a146e97464437eaaa19b37355d88fe : SUCCESS in 1h 31m 31s","accounts_in_message":[],"_revision_number":3},{"id":"5cb56108cfa7afbe4a2ba26f1610b8aaa60436c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 11:43:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"fc824ffcd7689d1f22a35363f00937b2b2cccf4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 11:45:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0d440880801447783d70027f4d1dc98 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9038a5baddeae8400c7fb04a4a2ab03a38c06ad8","revisions":{"3d147703c6e1acebb572ce6a05f8888d2f05c64d":{"kind":"REWORK","_number":1,"created":"2022-02-05 10:48:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/827947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/47/827947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/47/827947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/47/827947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/47/827947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/47/827947/1"}}},"commit":{"parents":[{"commit":"34e1446f44d66a96c8fbe631fc29413de0a51551","subject":"Add pymemcache\u0027s HashClient retrying options to oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/34e1446f44d66a96c8fbe631fc29413de0a51551"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-05 10:47:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-05 10:47:15.000000000","tz":540},"subject":"privsep: Allow customizing section name","message":"privsep: Allow customizing section name\n\nSome components uses sections not following the current name template\n(privsep_${section}) to register oslo.privsep parameters. For example\nnova registers the parameters to the nova_sys_admin section.\n\nThis change allows overriding the section name to deal with such cases.\n\nChange-Id: Icaf88ebaaf72d6810d9ded119a9998538eb09869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3d147703c6e1acebb572ce6a05f8888d2f05c64d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3d147703c6e1acebb572ce6a05f8888d2f05c64d"}]},"branch":"refs/heads/master"},"b6397d34f8f6b2da04a9b787ef9c4ef8f2025eea":{"kind":"REWORK","_number":2,"created":"2022-02-05 10:53:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/827947/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/47/827947/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/47/827947/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/47/827947/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/47/827947/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/47/827947/2"}}},"commit":{"parents":[{"commit":"34e1446f44d66a96c8fbe631fc29413de0a51551","subject":"Add pymemcache\u0027s HashClient retrying options to oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/34e1446f44d66a96c8fbe631fc29413de0a51551"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-05 10:47:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-05 10:52:48.000000000","tz":540},"subject":"privsep: Allow customizing section name","message":"privsep: Allow customizing section name\n\nSome components uses sections not following the current name template\n(privsep_${section}) to register oslo.privsep parameters. For example\nnova registers the parameters to the nova_sys_admin section.\n\nThis change allows overriding the section name to deal with such cases.\n\nChange-Id: Icaf88ebaaf72d6810d9ded119a9998538eb09869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b6397d34f8f6b2da04a9b787ef9c4ef8f2025eea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b6397d34f8f6b2da04a9b787ef9c4ef8f2025eea"}]},"branch":"refs/heads/master"},"9038a5baddeae8400c7fb04a4a2ab03a38c06ad8":{"kind":"REWORK","_number":3,"created":"2022-02-07 10:42:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/827947/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/47/827947/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/47/827947/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/47/827947/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/47/827947/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/47/827947/3"}}},"commit":{"parents":[{"commit":"34e1446f44d66a96c8fbe631fc29413de0a51551","subject":"Add pymemcache\u0027s HashClient retrying options to oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/34e1446f44d66a96c8fbe631fc29413de0a51551"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-05 10:47:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-07 10:42:33.000000000","tz":0},"subject":"privsep: Allow customizing section name","message":"privsep: Allow customizing section name\n\nSome components uses sections not following the current name template\n(privsep_${section}) to register oslo.privsep parameters. For example\nnova registers the parameters to the nova_sys_admin section.\n\nThis change allows overriding the section name to deal with such cases.\n\nChange-Id: Icaf88ebaaf72d6810d9ded119a9998538eb09869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9038a5baddeae8400c7fb04a4a2ab03a38c06ad8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9038a5baddeae8400c7fb04a4a2ab03a38c06ad8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
