)]}'
{"id":"openstack%2Fpuppet-oslo~830599","triplet_id":"openstack%2Fpuppet-oslo~master~I2acdf52bf0bc1c68037bb5326707a2eb3e3834a9","project":"openstack/puppet-oslo","branch":"master","topic":"use_tpool","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-08 01:49:08.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I2acdf52bf0bc1c68037bb5326707a2eb3e3834a9","subject":"db: Deprecate support for use_tpool","status":"MERGED","created":"2022-02-23 11:51:05.000000000","updated":"2022-03-08 06:39:22.000000000","submitted":"2022-03-08 06:37:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830599-use_tpool","meta_rev_id":"da451e85a39672b600d86caf4cb5c521dd92f00f","_number":830599,"virtual_id_number":830599,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-08 06:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-02-24 09:12:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-03-07 23:42:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-03-07 23:42:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-23 13:04:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-24 09:12:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-03-07 23:42:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"3666065f28f422bfffdfa6514dad7b439d7060a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 11:51:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4154cfa91e38e4b384450e5a54da9dff8302d92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 13:04:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef384096a08445f5bf11199bc686a983 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ee6787c778f46d7924b1e965d945e11 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e24d21be6c3842bf972de48686025b51 : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa0123d2afa54405b46c55e0fb55ad67 : FAILURE in 13m 46s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ce57edb28de4217833fe2d6d8cad911 : FAILURE in 11m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cedcb546103f4b4ca15027699a98184e : FAILURE in 7m 53s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/355ebfb03e5b4dc48834d5a333172f01 : FAILURE in 8m 32s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd439414d2484008a49e70b866c00784 : FAILURE in 8m 50s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/783438f5fda545aaae03bb1aafd1c137 : FAILURE in 8m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b5022b64635440dac3617f8050a17ae : FAILURE in 7m 32s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47e23e99292542378e3f13e7519b5728 : FAILURE in 7m 08s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b8221465e7f4225b1af8c0fbbe080bb : FAILURE in 7m 58s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad68278ea17a4db5a4346237ad2f0261 : FAILURE in 11m 51s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/141ad9a96bd24a62bf8c0f03b93fe876 : FAILURE in 11m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0c253e8212a471bb7e2109d39503a8a : FAILURE in 12m 22s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2606a42c691340c2ac8567cfd3a1ff4c : FAILURE in 11m 58s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3656de4c121c4e20bf16cb95d1544aa0 : FAILURE in 10m 32s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3e73699be314d89abbffc4709fdfb6e : FAILURE in 11m 18s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/57c17d15a4b545ee914d420beaf0fd4e : FAILURE in 14m 22s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/545ff8645eef421aa7cf49dd42264d7a : FAILURE in 17m 58s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cabefa5e7264474a9639820bdc137b89 : FAILURE in 9m 56s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7f4cc890c2742dfbd66d6bd6d9deb7b : FAILURE in 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/585c21b42ecc42c7837f03149984fc3d : SUCCESS in 3m 08s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/731d6d5db69f44ca9ea0de73748e0373 : FAILURE in 1h 00m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d78ee8e851f6207a66fb5555338655ad9a035291","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 13:25:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"aa0b0682611897912761ec7b4610f795f3151d69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 14:03:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3d295b7be629033b59cd3dea0d89f6326cd5c055","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 14:56:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3f842d1c6adefa0a6e47e912a9b1f99ee2daa669","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 16:45:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9135ca3c8abe46718c8926c95204792e : SUCCESS in 3m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dce8090835d4bea8f551a3969cc0a40 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/944aab3af0e64c48bf28f8e1ac341b27 : SUCCESS in 3m 35s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/21c7ca728a7540ddb20c7f60bc512a60 : SUCCESS in 9m 32s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/604a752b98f641b8b630c705fc7036fd : SUCCESS in 10m 59s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4381642b1df84987ac012815cd1d5b79 : SUCCESS in 8m 39s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acb10922966f423dbd9a570831f31173 : SUCCESS in 8m 54s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd68ead2b19341b39c1dc95d70807f19 : SUCCESS in 8m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec5ac44cf25d4fab8a7925ce48e1bb2b : FAILURE in 46m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/903245a62a3f4909b9a7b70370a18de9 : FAILURE in 36m 06s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a600f9f8a7a4d62a2eb619563138a0e : FAILURE in 49m 40s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8302d647053c4afcbe2e1e09d881a1ce : FAILURE in 28m 44s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4635b022cae4d7e917d6b60ca35b145 : SUCCESS in 21m 43s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/def8090cf1af4f5392aaf393c46ee430 : SUCCESS in 1h 15m 06s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9cedb0d758514afe834fe32f258cbb2a : SUCCESS in 1h 21m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/336820d9eadb4fbcaaf83d3e57cc2689 : SUCCESS in 1h 21m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c09b5a4d519f4d9b9489012a722bc31e : SUCCESS in 58m 41s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7fbc1b4292ab4e7b8df5905063dcd8b2 : SUCCESS in 20m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c55e3d30ada8433186e5905a56f2b929 : SUCCESS in 45m 34s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb6b3bfdff914cce96ecdcc85d435e7a : SUCCESS in 53m 51s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc67a0f183c5457fbe2cde248cd236df : SUCCESS in 1h 01m 18s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b62210b82fa41a78c3c8e3247ca191a : SUCCESS in 43m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a073f86936d14efb8d1016e49baee847 : SUCCESS in 3m 57s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/42c4e48b994145c58ebb888e359dc5a8 : SUCCESS in 1h 22m 02s","accounts_in_message":[],"_revision_number":4},{"id":"32379e3a20ca157d2cefcb26355f04f30e0f96a4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-24 09:12:55.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3c2cb2c7c5ecd712dc99cdbc85791f9f61fd671f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-07 23:42:23.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e418bac1cb0bd90c087e7fca9f9e10957e3cc84d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 23:44:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"15d26a6acc2633f0648e4c86fb6770c9c6df944e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 01:13:44.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87ac7fa004ec4ae0a71cbf83ad081b0b : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/741bdd7e34dd4c2abe6897af9b101591 : SUCCESS in 5m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a871ec05f811472081ab3d88657c6d14 : SUCCESS in 4m 33s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9151d5fd244746d9ae71da6c5b809d49 : SUCCESS in 13m 14s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a801a06192ed414688013785622d2bbf : SUCCESS in 10m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/112f81c215be4fc59c04e89e7fce3eab : SUCCESS in 11m 23s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e616fe8f03645c283a30351ab39eac8 : SUCCESS in 10m 26s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2645d9103f1048f1b6434aee100b9f9a : SUCCESS in 1h 06m 16s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ada0b7afc11d4c11a092ca6b35d488b3 : FAILURE in 56m 51s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c2856a3c00a540e5ab8f2d0404442549 : SUCCESS in 1h 10m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ecfee6a25e6a41beba96121fbcfcc9d7 : SUCCESS in 1h 06m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfb43898fe7642afa230fd8c23c8a6c1 : SUCCESS in 5m 10s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/41056f2e37a94ab090b4fc0d6251b2ae : SUCCESS in 1h 20m 28s","accounts_in_message":[],"_revision_number":4},{"id":"37e3d908ee10321987c3b6e43d46933fe5b2d653","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-08 01:49:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"74d6b5cc1f0e8cdc27489b6bdb4ddc966ab55678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 04:18:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3baf8191dcf24902a04f1e9274844599 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9100e0b5e79f4092a7055b0f48eb782c : SUCCESS in 4m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46c8bff36df14ba482413b923e65ed84 : SUCCESS in 4m 39s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3717c37d56f44e86b5ae9a224e243fc1 : SUCCESS in 11m 19s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/184a9f3d7dd54020847172b36ec96be7 : SUCCESS in 12m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adc923a00c624f11b1d847a31cba60ee : SUCCESS in 9m 52s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25e995363e944d529ba7d86bfff4144b : SUCCESS in 8m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6837538e418c4d39aa5b4172f6816c36 : SUCCESS in 8m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c2b65e1c63e4410af857ee6c87f97d3 : FAILURE in 41m 31s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa475e26f1c94f89ba9b0675b0d37cd4 : FAILURE in 47m 07s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3cc663df4984e7f96d5b014a8c0b5f9 : FAILURE in 41m 17s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae2fc21dd4424f02a2e6209e0b71ff25 : FAILURE in 38m 30s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bfbfe66fa66c45f7ad209286601a1e49 : SUCCESS in 23m 51s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9545f1bff9b1473aa26b7e03c79f3563 : SUCCESS in 1h 10m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b970f20fd59c4595ae0a6c87c5deb771 : SUCCESS in 1h 24m 24s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ff500aa9992d4a94b5981112785415d5 : SUCCESS in 1h 22m 18s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1170de77961c4135a475312009f61310 : SUCCESS in 1h 08m 42s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c7bf6f2e4334392a534d31bd5ab562d : SUCCESS in 22m 18s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1683d9e27aa84ad3a369b5e3109dc9a0 : SUCCESS in 1h 03m 06s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0f30325cd9a428a974381328a2d42a2 : SUCCESS in 1h 05m 33s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/01a36ec89b9c4050a97e5e0b082b6ff6 : SUCCESS in 1h 05m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af5bb83ebe864545abed8c95eaa8e93a : SUCCESS in 54m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0788ade7f8e7428885d3b959b7a002ad : SUCCESS in 4m 41s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/18df2fca8bed4c4bb764a83062d5df51 : SUCCESS in 1h 31m 10s","accounts_in_message":[],"_revision_number":4},{"id":"3445766976b788427bb371247ebd6131dc21d360","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 04:20:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3348638136ee85955a2c1430277d16a69c22ae3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 06:37:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61139cafcc4b4442a12a962d88bbf131 : SUCCESS in 4m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68b36a00a7364e69a12aa133c3963b95 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e73ca6f46b2e4469a9f2eec1097fbcf2 : SUCCESS in 4m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1dbcaf1d32a548a1b97c00fe75340bc1 : SUCCESS in 12m 14s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/509603ee55a64f8db76af848cd03de4d : SUCCESS in 10m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db53090d265142c5bcf39f3f8a56f67f : SUCCESS in 8m 46s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26286b34502b4bdab738ddc2e60dc469 : SUCCESS in 8m 49s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/633f996b1c434ae083a76e4dd8fcc92c : SUCCESS in 1h 10m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7178c27c034c49458c99b7932e74dc51 : SUCCESS in 1h 16m 47s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/86bfd4d84a7c44c1a196da85effdff8f : SUCCESS in 1h 06m 29s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e5edd991a673480d9c75f8944d12e14e : SUCCESS in 57m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b69feaf7c94842f28b3976773aa7f2f5 : SUCCESS in 5m 09s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/419d8b7bf5aa4d2a96f5b784f0cbaed0 : SUCCESS in 1h 37m 46s","accounts_in_message":[],"_revision_number":4},{"id":"ada73d862a9c751dbe5ac8f600d5f49d60540aca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 06:37:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"da451e85a39672b600d86caf4cb5c521dd92f00f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 06:39:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c54a0c5f07646c78ad9cbd825808cbf : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"af6514829c48a4a6c06d06abc66552e4424fdc81","revisions":{"d9e60543d7bbd11eb54fa1a3e7ef40f58c8916cf":{"kind":"REWORK","_number":1,"created":"2022-02-23 11:51:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/830599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/99/830599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/1"}}},"commit":{"parents":[{"commit":"b299c0d88fb00f44522d23803333631c616d0687","subject":"Merge \"privsep: Allow customizing section name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b299c0d88fb00f44522d23803333631c616d0687"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 11:49:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 11:50:56.000000000","tz":540},"subject":"db: Deprecate support for use_tpool","message":"db: Deprecate support for use_tpool\n\n... because it was deprecated during Xena cycle[1].\n\n[1] 74c6bf266e0b86ceb4726cad9268f134d423bae8\n\nChange-Id: I2acdf52bf0bc1c68037bb5326707a2eb3e3834a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d9e60543d7bbd11eb54fa1a3e7ef40f58c8916cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d9e60543d7bbd11eb54fa1a3e7ef40f58c8916cf"}]},"branch":"refs/heads/master"},"4ed46c58b2d28f3c237c27ec63c7b68a7ba7596c":{"kind":"REWORK","_number":2,"created":"2022-02-23 13:25:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/830599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/99/830599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/2"}}},"commit":{"parents":[{"commit":"b299c0d88fb00f44522d23803333631c616d0687","subject":"Merge \"privsep: Allow customizing section name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b299c0d88fb00f44522d23803333631c616d0687"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 11:49:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 13:25:10.000000000","tz":0},"subject":"db: Deprecate support for use_tpool","message":"db: Deprecate support for use_tpool\n\n... because it was deprecated during Xena cycle[1].\n\n[1] 74c6bf266e0b86ceb4726cad9268f134d423bae8\n\nChange-Id: I2acdf52bf0bc1c68037bb5326707a2eb3e3834a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/4ed46c58b2d28f3c237c27ec63c7b68a7ba7596c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/4ed46c58b2d28f3c237c27ec63c7b68a7ba7596c"}]},"branch":"refs/heads/master"},"46c342092be14088db3a11ca5a6e6fd6ce908792":{"kind":"REWORK","_number":3,"created":"2022-02-23 14:03:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/830599/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/99/830599/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/3"}}},"commit":{"parents":[{"commit":"fe04834873846952df0f854b4c0a78901cb2e3a2","subject":"CentOS 9: Disable unit tests dependent on puppet-postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/fe04834873846952df0f854b4c0a78901cb2e3a2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 11:49:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:03:27.000000000","tz":540},"subject":"db: Deprecate support for use_tpool","message":"db: Deprecate support for use_tpool\n\n... because it was deprecated during Xena cycle[1].\n\n[1] 74c6bf266e0b86ceb4726cad9268f134d423bae8\n\nChange-Id: I2acdf52bf0bc1c68037bb5326707a2eb3e3834a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/46c342092be14088db3a11ca5a6e6fd6ce908792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/46c342092be14088db3a11ca5a6e6fd6ce908792"}]},"branch":"refs/heads/master"},"af6514829c48a4a6c06d06abc66552e4424fdc81":{"kind":"REWORK","_number":4,"created":"2022-02-23 14:56:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/830599/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/99/830599/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/99/830599/4"}}},"commit":{"parents":[{"commit":"fe04834873846952df0f854b4c0a78901cb2e3a2","subject":"CentOS 9: Disable unit tests dependent on puppet-postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/fe04834873846952df0f854b4c0a78901cb2e3a2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 11:49:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:36:52.000000000","tz":540},"subject":"db: Deprecate support for use_tpool","message":"db: Deprecate support for use_tpool\n\n... because it was deprecated during Xena cycle[1].\n\n[1] 74c6bf266e0b86ceb4726cad9268f134d423bae8\n\nChange-Id: I2acdf52bf0bc1c68037bb5326707a2eb3e3834a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/af6514829c48a4a6c06d06abc66552e4424fdc81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/af6514829c48a4a6c06d06abc66552e4424fdc81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
