)]}'
{"id":"openstack%2Fpuppet-oslo~831352","triplet_id":"openstack%2Fpuppet-oslo~stable%2Fwallaby~Ifdf513a8202ba81938b1e9224c430d9a15192556","project":"openstack/puppet-oslo","branch":"stable/wallaby","topic":"c9_wallaby_gates","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-03-03 09:07:09.000000000","reason":"Marios Andreou replied on the change"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-02 11:42:08.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"Ifdf513a8202ba81938b1e9224c430d9a15192556","subject":"Add centos-9 jobs for wallaby zuul layout","status":"MERGED","created":"2022-03-01 10:04:09.000000000","updated":"2022-03-04 11:06:36.000000000","submitted":"2022-03-04 11:06:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831352-c9_wallaby_gates","meta_rev_id":"b82364f59a5f510cba5c0f8784553ff7246818b5","_number":831352,"virtual_id_number":831352,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-04 11:06:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2022-03-04 09:06:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2022-03-02 11:42:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2022-03-04 09:06:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-01 12:10:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 09:43:02.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"5d73e26642e2099167a76cd91a6092b14e565fef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-01 10:04:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f33d5021269237779eb0a22f33a9965628b12a27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 12:10:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/369a316cd5f04434b0d8793abc80094d : SUCCESS in 5m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/028073f0126e4130a6fa3715a66c56dd : SUCCESS in 5m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bedfd899a482415d8b09897d30b42d39 : SUCCESS in 12m 02s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49e419b7c92541ce8003ac1343f5fd1f : SUCCESS in 10m 35s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89b9f343859c4c23a33c815f96f513db : SUCCESS in 9m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c7be2ace6f4451a9ac004d6ac753c89 : FAILURE in 38m 42s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c4624b2c7c549c99f5e3825d388cd5f : FAILURE in 1h 14m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6e12c1c8e28478189e303adf758de97 : FAILURE in 35m 09s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/549ae0b6d4e7419f9f53e2cf5b22f924 : FAILURE in 30m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/616fa517a9c9426093a15ff29313526e : SUCCESS in 1h 17m 43s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/885720425cab4de1b978e83b3223c54f : SUCCESS in 1h 32m 02s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6225992aa457404d81308e6083370fce : SUCCESS in 1h 16m 46s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/31d63b462cd84b378aede01d668237a2 : SUCCESS in 52m 18s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/05c4d76f49b54643b55dfa732b62a7f6 : SUCCESS in 2h 02m 26s\n- puppet-oslo-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/a6097f0297da4f1288019ea3f641f2d3 : SUCCESS in 1h 17m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78486ac8d1de82cb472f7362e3adc5622ed60e3c","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-01 14:57:00.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2535c82d646673d4873456baa2c69c73be87e403","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-01 15:14:31.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1e7504ce7f0fdf9f4ad7a26bb91c3af61a865e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 17:23:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ea6ed6bbbfa4ec784a50e0dbc90d8f5 : SUCCESS in 5m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e18e643957ef4e3b934cfb909462026d : SUCCESS in 5m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5bd4c5726f2421899e5539fdb131ee9 : SUCCESS in 13m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3df44af68ff84fabb09120fd9753211e : SUCCESS in 10m 36s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84f48e055a2d4660b4dc072b70c049b8 : SUCCESS in 10m 49s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd90a3f6e7f04859a4c500b6d5aae617 : FAILURE in 36m 46s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27cabb10771a4a67b65840715b43ad9e : FAILURE in 1h 14m 30s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbaf64d0b6424f5faa9c0731fbe80e51 : FAILURE in 41m 47s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72eaaa9235ae4465a5fec9c7623eb2d1 : FAILURE in 27m 47s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d840f573b084962afb9d03732d96999 : SUCCESS in 1h 17m 11s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6f0da934ea24f8f9e05f2c69edb1ff0 : SUCCESS in 1h 28m 34s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/59f5db9ae7cf4bd8a700964f8e1a4ddd : SUCCESS in 1h 19m 41s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4cc5ab2ca28f416dbb05d5cc24a49fba : SUCCESS in 1h 07m 04s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a31a8913f4fd403884c8f721aecb14f8 : SUCCESS in 2h 01m 46s\n- puppet-oslo-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/3656dbe861f3408c9fc295ba312efbfb : SUCCESS in 1h 36m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8a8f8c4d450692cddc640423a7305dfc67f144d","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-02 09:42:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4160c00dea5f0b08ba6924750281e04947823fb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-02 09:42:37.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2c59109621c79b34cce82d9508e85db7311c765a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-02 11:42:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b154a68dac877e309f6f4e0d279e4e9c1b4854eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 12:05:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3c8b3f5318e4ae28e2e2d3fceed5e69 : SUCCESS in 5m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f8f01448e3148b2b021b09e00859b04 : SUCCESS in 5m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f7c2f34b92734a3ab2b8afe5a1ac96e3 : SUCCESS in 14m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de548697b77f48789299903bcf7340bf : SUCCESS in 11m 32s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06685794c6014392bb8654b1101705ad : SUCCESS in 8m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcf711361b97494eb335d6f5a258c9e1 : FAILURE in 48m 53s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/466e54c8b433405fb978e5aabbaaba73 : FAILURE in 1h 20m 00s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6424e8d16d554f13887994a8549f303a : FAILURE in 1h 08m 26s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ba635cd343e4504b253365786081a2b : FAILURE in 53m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8788b743cb3f4e54a6736de9a4446bc9 : SUCCESS in 1h 38m 14s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ff927c243e9642e6b352a58d9e00ab1a : SUCCESS in 1h 43m 56s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f9996d415ce4a2493181e7b9ee90764 : SUCCESS in 1h 33m 16s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/50f10d6747d64e43a38cc6d6ebed6c55 : SUCCESS in 1h 08m 11s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5dbcf35484b84425bf9c4294ad0ea94e : SUCCESS in 2h 15m 10s\n- puppet-oslo-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/56995338b78e4c9894f6bb371756072c : SUCCESS in 1h 21m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f340e49f37fbecd414ac12b12991678997a2314d","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-03 09:07:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ab7db9f9f8c36cd6e43511f7ecae3c8987575376","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-04 09:06:03.000000000","message":"Patch Set 3: Code-Review+1 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2a1ba1ad83eb637154c99a06966af420ebea2094","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 09:08:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9938ab7119e442bf9a8d27ac812244d8303c02b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 11:06:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d831c179285457492cdd28b43ba2a8b : SUCCESS in 5m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ac60b34248f43459768642b9081cf5c : SUCCESS in 5m 06s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a5693c7db0547a8a6b9a6841fcf24d9 : SUCCESS in 12m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/509b63b6b6134d91a6f2e7eb83f58772 : SUCCESS in 8m 01s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b919cb85cd44692ad002ac7312a9828 : SUCCESS in 1h 11m 34s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ea9c323f22248c89a4b25b0629bb854 : SUCCESS in 1h 28m 41s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8989a9c7be3c4abfb649e9f8d5afb4a6 : SUCCESS in 1h 22m 25s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bfa1b264ff944891ad049053e033e95a : SUCCESS in 1h 06m 07s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7fdb0827c9ec41169833ccb0f9b89d26 : SUCCESS in 1h 56m 59s","accounts_in_message":[],"_revision_number":3},{"id":"b82364f59a5f510cba5c0f8784553ff7246818b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 11:06:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"aa67229ad9171385f24e025fc01b89b0ea1ea0d4","revisions":{"2e9596dfbfade6cdc550d8da661da4feed46f039":{"kind":"REWORK","_number":1,"created":"2022-03-01 10:04:09.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/52/831352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/52/831352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/52/831352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/52/831352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/52/831352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/52/831352/1"}}},"commit":{"parents":[{"commit":"d80fbfc9de7e46c8c943454671810a08e43af6e5","subject":"Add pymemcache\u0027s HashClient retrying options to oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d80fbfc9de7e46c8c943454671810a08e43af6e5"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-03-01 10:01:37.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-03-01 10:01:37.000000000","tz":120},"subject":"Add centos-9 jobs for wallaby zuul layout","message":"Add centos-9 jobs for wallaby zuul layout\n\nAdds puppet-oslo-tripleo-centos-9-standalone definition to the\nzuul layout. This is added as non voting as puppet openstack\nwill not support c9 on wallaby.\n\nSee the topic branch [1] for related reviews and [2] for context.\n\n[1] https://review.opendev.org/q/topic:c9_wallaby_gates\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2022-February/027403.html\n\nDepends-On: https://review.opendev.org/c/openstack/puppet-openstack-integration/+/831206\nChange-Id: Ifdf513a8202ba81938b1e9224c430d9a15192556\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/2e9596dfbfade6cdc550d8da661da4feed46f039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/2e9596dfbfade6cdc550d8da661da4feed46f039"}]},"branch":"refs/heads/stable/wallaby"},"79055a717ce61b4557204f0c723c569d027b6c99":{"kind":"REWORK","_number":2,"created":"2022-03-01 15:14:31.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/52/831352/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/52/831352/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/52/831352/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/52/831352/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/52/831352/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/52/831352/2"}}},"commit":{"parents":[{"commit":"d80fbfc9de7e46c8c943454671810a08e43af6e5","subject":"Add pymemcache\u0027s HashClient retrying options to oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d80fbfc9de7e46c8c943454671810a08e43af6e5"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-03-01 10:01:37.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-03-01 15:14:31.000000000","tz":0},"subject":"Add centos-9 jobs for wallaby zuul layout","message":"Add centos-9 jobs for wallaby zuul layout\n\nAdds puppet-oslo-tripleo-centos-9-standalone definition to the\nzuul layout. This is added as non voting as puppet openstack\nwill not support c9 on wallaby.\n\nSee the topic branch [1] for related reviews and [2] for context.\n\n[1] https://review.opendev.org/q/topic:c9_wallaby_gates\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2022-February/027403.html\n\nDepends-On: https://review.opendev.org/c/openstack/puppet-openstack-integration/+/831206\nChange-Id: Ifdf513a8202ba81938b1e9224c430d9a15192556\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/79055a717ce61b4557204f0c723c569d027b6c99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/79055a717ce61b4557204f0c723c569d027b6c99"}]},"branch":"refs/heads/stable/wallaby"},"aa67229ad9171385f24e025fc01b89b0ea1ea0d4":{"kind":"REWORK","_number":3,"created":"2022-03-02 09:42:37.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/52/831352/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/52/831352/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/52/831352/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/52/831352/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/52/831352/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/52/831352/3"}}},"commit":{"parents":[{"commit":"d80fbfc9de7e46c8c943454671810a08e43af6e5","subject":"Add pymemcache\u0027s HashClient retrying options to oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d80fbfc9de7e46c8c943454671810a08e43af6e5"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-03-01 10:01:37.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-03-02 09:42:37.000000000","tz":0},"subject":"Add centos-9 jobs for wallaby zuul layout","message":"Add centos-9 jobs for wallaby zuul layout\n\nAdds puppet-oslo-tripleo-centos-9-standalone definition to the\nzuul layout. This is added as non voting as puppet openstack\nwill not support c9 on wallaby.\n\nSee the topic branch [1] for related reviews and [2] for context.\n\n[1] https://review.opendev.org/q/topic:c9_wallaby_gates\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2022-February/027403.html\n\nDepends-On: https://review.opendev.org/c/openstack/puppet-openstack-integration/+/831206\nChange-Id: Ifdf513a8202ba81938b1e9224c430d9a15192556\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/aa67229ad9171385f24e025fc01b89b0ea1ea0d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/aa67229ad9171385f24e025fc01b89b0ea1ea0d4"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
