)]}'
{"id":"openstack%2Fpuppet-oslo~832643","triplet_id":"openstack%2Fpuppet-oslo~stable%2Fvictoria~If4622f85e1e7f715805b205bfc3c16cb1e53b943","project":"openstack/puppet-oslo","branch":"stable/victoria","topic":"victoria","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-03-11 02:42:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4622f85e1e7f715805b205bfc3c16cb1e53b943","subject":"Add retrying options to oslo.cache","status":"MERGED","created":"2022-03-08 14:18:49.000000000","updated":"2022-03-11 02:45:30.000000000","submitted":"2022-03-11 02:42:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832643-victoria","meta_rev_id":"180780394784d6b7c9d3e282a7a15b21741af543","_number":832643,"virtual_id_number":832643,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-11 02:42:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-03-10 06:22:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-03-10 23:52:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-03-10 09:46:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-03-10 23:52:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-08 16:52:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-10 06:22:27.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-03-10 09:46:05.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2022-03-10 23:52:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"1d29e86b1942468b055e55da8026f4a23d7cbaa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-03-08 14:18:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f274f7dc8b8a513033afef90873fd08f31cd37f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-03-08 14:19:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8cbf72bfba4712b0d378a72d05701f49d5710513","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 16:52:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a873e84bdf2f4166b76ef35a1d14d797 : SUCCESS in 6m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ad5937a37e5456f9538c1150e517228 : SUCCESS in 4m 54s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/759181dc1d674cf3a6ac2e077d5cc183 : SUCCESS in 12m 05s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38b887f6321c44368f0b9f0215a6cef1 : SUCCESS in 10m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4732f2cebd55406288b1714abc7c79b5 : SUCCESS in 10m 50s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9059bea682e44eaf977bd4c06d258d44 : SUCCESS in 53m 37s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9231c3037c694a87b8f3e57498930f2f : SUCCESS in 55m 31s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ce350074fe24359835f1c2962cb7777 : SUCCESS in 48m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b633049f73974c69a1068b594b85e8be : SUCCESS in 42m 13s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/17e9b78c5e4f41c392c71c917f8b9afc : SUCCESS in 1h 13m 09s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/94104bbd26614de897cbc5406a49e821 : SUCCESS in 1h 20m 54s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/88299c36af9d45a285b11e6c241f2d7a : SUCCESS in 1h 26m 59s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6bed0b4d6dc74de883012db9a7ace78a : POST_FAILURE in 1h 46m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a6d00607ba34542ad9a44452afa07c3 : SUCCESS in 5m 41s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6182348e455a4d779833b0808d102f83 : SUCCESS in 2h 26m 52s","accounts_in_message":[],"_revision_number":1},{"id":"826835080e69743c0984beb22f4aad3b7e3f678d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-03-09 10:09:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c737baa1be9854430003a3ed0c61c6c5a0385236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 12:20:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8976392247f4d3684ca6b9b774b5cf0 : SUCCESS in 4m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5127efb293af4185821cddd4294cd695 : SUCCESS in 4m 51s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ffac8dc3100439ba385baa88aa32a04 : SUCCESS in 11m 17s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac02099a99d4406f85306d2aec32c0d6 : SUCCESS in 8m 47s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b073205f2e0f430bacc8e084f59e3acc : SUCCESS in 9m 01s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6f2dcd6a9054c2c8137ebb19e4363b9 : SUCCESS in 45m 04s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f9d8f1063f144c887b44c3a9496b3f1 : SUCCESS in 49m 09s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7f6c22ff6bd4493878b51f087b3d4ec : SUCCESS in 49m 00s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12df32af1c9a4c209313afbc1772e7f8 : SUCCESS in 37m 31s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac2f4c308d8a41be8f1553e2784d1737 : SUCCESS in 1h 08m 16s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3915327adee3459e9567c23769b702a8 : SUCCESS in 1h 20m 31s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/92875595e2cd4be08ce382d98a4744ff : SUCCESS in 1h 20m 50s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/03bd92d86325453a8122ccd567572b4b : SUCCESS in 56m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ee3d123fa484bdaafd511a6f685945d : SUCCESS in 5m 15s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6b7542411e894667940d3564e39568ca : SUCCESS in 2h 05m 33s","accounts_in_message":[],"_revision_number":1},{"id":"69b94ccf47f44b0770e28dfdd578de11aadfe40b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-10 06:22:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"de9a39a7d3ea830d2677248585bd757d7c6feae2","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-03-10 09:46:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"321a48f61fbe867fecabaf3e3b90170c807d023c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-10 23:52:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37a39edf29cfdc9d83dbdb788af77865ef29460f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 23:54:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02d045902f315a35df18c38b051b340a8e48b953","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 02:42:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2feb4d8a2df64c759e5f647109eb4b60 : SUCCESS in 5m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df80fcaac20d4c5f8659354e6276b5bd : SUCCESS in 5m 11s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/6cd09676401343ac892accd6df4151db : SUCCESS in 11m 15s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47037772718a4fbda9c1533e43399eac : SUCCESS in 8m 20s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/076f45f1111040aaa05970e1637b9ad5 : SUCCESS in 1h 23m 50s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d6f3761dc47b447d9156acd86be1a1ba : SUCCESS in 1h 52m 53s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0226654c84984d3da279ea7c3a972a2b : SUCCESS in 1h 34m 26s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca214a593bb64a859d300b17991b00a3 : SUCCESS in 1h 07m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a0f20adceb4484ab1f373311127d6c5 : SUCCESS in 5m 21s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c18a3e11986f424681175be3438d6c46 : SUCCESS in 2h 42m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ab78b8d8167c1cf03a5d1809dcaaa84704ac8ecb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 02:42:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"180780394784d6b7c9d3e282a7a15b21741af543","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 02:45:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff908649e3d94dd8bc9a3991be1c2732 : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c62c4bccb80dde5ef4c4b050a75d5024bf52b5a0","revisions":{"c62c4bccb80dde5ef4c4b050a75d5024bf52b5a0":{"kind":"REWORK","_number":1,"created":"2022-03-08 14:18:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/43/832643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/43/832643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/43/832643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/43/832643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/43/832643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/43/832643/1"}}},"commit":{"parents":[{"commit":"dc42ef8ed7b5804491bd1355fd8bdeb84b4ca629","subject":"Add socket keepalive options for the pymemcache backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/dc42ef8ed7b5804491bd1355fd8bdeb84b4ca629"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-01-28 10:24:04.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-03-08 14:18:20.000000000","tz":60},"subject":"Add retrying options to oslo.cache","message":"Add retrying options to oslo.cache\n\nThis patch specifies a set of options required to setup the retrying wrapper\nfeature of pymemcache (dogpile.cache) cache backend.\n\nOriginal oslo.cache change:\nhttps://review.opendev.org/c/openstack/oslo.cache/+/803747\n\nChange-Id: If4622f85e1e7f715805b205bfc3c16cb1e53b943\n(cherry picked from commit f7dbe8aa5f4ebc31d77344932fef2eea84a32906)\n(cherry picked from commit 88197f306123c521967b5a16b4909e5227554325)\n(cherry picked from commit c6fe2e52e3aac5722c5a1f855951017543a8a704)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c62c4bccb80dde5ef4c4b050a75d5024bf52b5a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c62c4bccb80dde5ef4c4b050a75d5024bf52b5a0"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
