)]}'
{"id":"openstack%2Fpuppet-oslo~834200","triplet_id":"openstack%2Fpuppet-oslo~master~I48b5b4d556dad468a031a84e598223aae3ae7cb5","project":"openstack/puppet-oslo","branch":"master","attention_set":{},"removed_from_attention_set":{"6926":{"account":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"last_update":"2022-03-18 11:00:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I48b5b4d556dad468a031a84e598223aae3ae7cb5","subject":"Allow merging default log levels for Oslo log","status":"ABANDONED","created":"2022-03-17 16:44:11.000000000","updated":"2022-03-18 14:58:24.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"716796ab3ebeda02c3981f370956155cdb17eb8c","_number":834200,"virtual_id_number":834200,"owner":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-17 17:57:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-03-17 16:44:49.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-17 17:16:59.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-03-17 17:57:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0a3f2f986dc2aa4afd723f59d1bf7f5145b890fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-03-17 16:44:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d48363ba43903ff293f9807b11b344de48ee1bc7","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-03-17 16:44:49.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"446b7e51e7ab87e4c6ecc0f39d9610f77529158c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-17 17:16:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"056710871b4afe64c9250aa5ee908d3e40d8d7cc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-17 17:22:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"316740ac3acf6cd4e2398b1d5dd0bdf0e59e9dd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 17:57:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/375ccdaf79b746ea94469e2e44e7654b : SUCCESS in 5m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da899669e0434620b95131ddd1340215 : FAILURE in 5m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/996bf70c8a4943739db121c144cc6e20 : FAILURE in 5m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba6fc15f1d59401797a89a1f721b8e01 : FAILURE in 12m 25s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e57331a193a4672a5f69eb440746f0d : FAILURE in 12m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bf9c2e4502e46b29396d5f388e5ae81 : FAILURE in 12m 53s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e526b39d7384aed90a26b999f97253d : FAILURE in 12m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0efe3ef3d26f427180b8e99468f4015a : FAILURE in 12m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01c8c7e801f2404aa73cae8dd1845a19 : FAILURE in 13m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9aa4eeb4a0774389bba006e2644439f8 : FAILURE in 13m 43s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acefc2c5c4f645fe91df5e7844f2852a : FAILURE in 12m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8df403db3f714a5aad1757a195f3d104 : FAILURE in 12m 33s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/49b257aa848f427ebc00f46c5258d973 : FAILURE in 19m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fac08f908f894a15bbd53fd8664ac792 : FAILURE in 17m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/66f2b24d18f64e57ad5e931d0ef404b1 : FAILURE in 17m 37s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa75f60deb444bc9bb35df048548dee6 : FAILURE in 24m 14s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0661566c1bef45ecb243c9300ecdf45c : FAILURE in 17m 34s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1549b21cbd54231847d65135e7f3277 : FAILURE in 16m 49s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/534dbddfa5dd4e0787f98778f55c7e9a : FAILURE in 16m 30s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7468cc6b6764bb5979ac81ce7108037 : FAILURE in 15m 54s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe9daa0ddfa44af2b318948eab98abf1 : FAILURE in 17m 28s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f3c5e7897774106b473435cf9c29327 : FAILURE in 16m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba4e9de1559c45bb93d9ea0f02bb8fe9 : SUCCESS in 5m 56s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1b5593a5d199480b8b2265bb97a58861 : FAILURE in 1h 09m 31s","accounts_in_message":[],"_revision_number":1},{"id":"21f4bce21a3a7a01b824ee2bd6e886a669070252","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-03-18 11:00:24.000000000","message":"Abandoned\n\nmakes sense, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"716796ab3ebeda02c3981f370956155cdb17eb8c","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-03-18 14:58:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"897867fd77a3a90ccd7141d94997c5424046feca","revisions":{"897867fd77a3a90ccd7141d94997c5424046feca":{"kind":"REWORK","_number":1,"created":"2022-03-17 16:44:11.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/00/834200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/00/834200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/00/834200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/00/834200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/00/834200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/00/834200/1"}}},"commit":{"parents":[{"commit":"37185e7c5690e09f54a7ecbbca46857aa7eb8fa9","subject":"Prepare Yoga RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/37185e7c5690e09f54a7ecbbca46857aa7eb8fa9"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2022-03-17 16:41:56.000000000","tz":60},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2022-03-17 16:44:09.000000000","tz":60},"subject":"Allow merging default log levels for Oslo log","message":"Allow merging default log levels for Oslo log\n\nAdd merge_default_log_levels flag that merges default_log_levels\nwith the Oslo.log service defaults value for it. If there is no\noslo log library available in the python path, or the merging flag was\nnet set, default_log_levels would overwrite $::os_service_default\nfor Oslo log service (default behavior). Defaults to False.\n\nWith that flag set, users can modify log levels scope of a particular\nopenstack component without loosing the other services defaults.\n\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\nChange-Id: I48b5b4d556dad468a031a84e598223aae3ae7cb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/897867fd77a3a90ccd7141d94997c5424046feca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/897867fd77a3a90ccd7141d94997c5424046feca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
