)]}'
{"id":"openstack%2Fpuppet-oslo~834274","triplet_id":"openstack%2Fpuppet-oslo~master~I8c34677059c2989f7700f27443c91c807d76160c","project":"openstack/puppet-oslo","branch":"master","topic":"fix-request_id","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-04-21 07:53:27.000000000","reason":"Change was abandoned"},"6926":{"account":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"last_update":"2022-04-21 07:53:27.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8c34677059c2989f7700f27443c91c807d76160c","subject":"Add param to log extra context values","status":"ABANDONED","created":"2022-03-18 11:03:42.000000000","updated":"2022-04-21 07:53:27.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"253d397b792d3261598a7f24c41ffc31e6f578e2","_number":834274,"virtual_id_number":834274,"owner":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2022-04-11 18:39:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-04-11 18:38:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-04-11 14:11:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"date":"2022-04-11 18:39:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-18 12:30:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-08 14:24:42.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-04-11 14:11:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-04-11 18:38:58.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"25371e388fe4b98088ebafc720443f881ef7a908","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-03-18 11:03:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f2e6b7fdf3f3b4f7a304d7b3f78456a3232dfe8","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-03-18 11:03:53.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b22ca394ab7948a65e820350ea08be2f7b83c563","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 12:30:02.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"88502d82ba00844c53f27d32a261f168e5900761","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-03-22 09:39:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"044ad7ea709d52341bc5b59bbf66ec0298dc71d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 11:33:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9de1a286ffc84ce4bd656c706bdcf6df : SUCCESS in 4m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/764ebe8d9a29436e8ebec2475365f339 : SUCCESS in 5m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b30430193e34297a8926ed413216b0d : SUCCESS in 4m 47s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b52acc9fa87466eb235eb68186212d6 : SUCCESS in 11m 51s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb0964133f434369bc1766bffb8ae39f : SUCCESS in 15m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f23fff2a88c43adbc186b75438a1617 : SUCCESS in 9m 05s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0a864dcaa704ce8961447e431becd4b : SUCCESS in 9m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4340e9db66684547bffba59495f8e6bf : SUCCESS in 9m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8fe540efd8848f694b8b02a59e1ec5f : FAILURE in 32m 40s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d593e23b77e4713baf7a001f264b805 : FAILURE in 39m 35s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea731fe2f6b2449495063baadcf577f7 : FAILURE in 39m 19s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1471a39eefc4941ba2af162f458f04c : FAILURE in 51m 18s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a1140f659b54452b93c5f1932d2241d : SUCCESS in 21m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/966a794ac1f64c92b785e672971911c2 : SUCCESS in 1h 01m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/27742d37257f495b81b1d60cba134924 : FAILURE in 1h 11m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0296e39ed2404a068650021b4c70e395 : SUCCESS in 1h 05m 19s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1caa1533967346f79ee9fd3f91a20dd1 : SUCCESS in 1h 00m 35s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/20519c6bdac9451780b05c5d69474eeb : SUCCESS in 20m 59s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/770c15824d9342e59eed9211259d0b0d : SUCCESS in 1h 01m 48s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/392ac9d3540546b2ad4c7a3776ff81d7 : SUCCESS in 1h 03m 21s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/889b99d9af504685b2b9df7b393fe788 : SUCCESS in 1h 04m 09s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/63c1c10b306543f383d22e965a92909e : SUCCESS in 58m 53s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e31cccd02b084e6cbbcc10b561f70e68 : SUCCESS in 1h 42m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d663469916cc4c288b9e5415352548cd64752e87","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-03-30 08:33:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"000dbbc1be1fbdb74f3568bc3733fc7a4bb62313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 10:49:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/459f954a979146ab811d3d2f1e4161ee : SUCCESS in 4m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e02fffdfed8e4516906a465baaa19820 : SUCCESS in 4m 40s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ff3433f946e478db2bef0a7ada24854 : SUCCESS in 4m 47s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/93459eb19cca42f4bfe325712e9e39ae : SUCCESS in 12m 55s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/13978807608f49eda58670e8b81c62c2 : SUCCESS in 11m 56s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a080237696a649638c295a4165fa86d4 : SUCCESS in 9m 01s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd26c2fbadb44d0a95aa21a3abee3a19 : SUCCESS in 8m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a3dbf5975fe42d8a1a7ef91f2d36573 : SUCCESS in 8m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fa73d76972f4a43af0a912e6f701292 : FAILURE in 33m 50s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e64ef3cbc49142fbb8c3b60689a4f297 : FAILURE in 40m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62364fedabe6462da54a7458bd5cd947 : FAILURE in 42m 20s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2306eeef7c404d878e6ab73e0a15a5cf : FAILURE in 38m 12s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cce34ee124c34eb8b17722084ed0b92c : SUCCESS in 26m 02s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3df0301da3384533b7d079884467fd3f : SUCCESS in 1h 04m 37s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a179b6feae242fdb7c8765085b11dc8 : SUCCESS in 1h 14m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d8a1fdf3a65460eb511fb2aa4b59067 : SUCCESS in 1h 10m 22s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b8332e24ce54c01bf8e0e505ab3ec26 : SUCCESS in 1h 04m 20s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0166c4c289ce499d9f54c3bb7d06090a : SUCCESS in 20m 14s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3aa389e14e643de94d280b9657515d8 : SUCCESS in 49m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bea2ae4ab4e427790aa421ed87df279 : SUCCESS in 1h 01m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6c16f7de0db47208a82246a1bc04b2c : SUCCESS in 1h 50m 18s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/047b892caa8f489b8bcade6ef28e4024 : SUCCESS in 1h 00m 24s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/891c9ed2d1c4418790813e31c2b6ba09 : SUCCESS in 1h 22m 53s","accounts_in_message":[],"_revision_number":1},{"id":"107c9e81f5a720aee8be6017ce29e7ece2845c48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-04-08 14:14:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"adf64d58f85c06dabbc564570bbed838d7f4fa81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-04-08 14:23:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"33ce0eff79459f589c1661e2431d42f5c8c5b727","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-04-08 14:27:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"89e886361dbdfc5b0b9995835303958be00887ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 16:05:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6db8853c240c41c4972668979d71144a : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11beed0620a140b6b6aecc48c2c0b7dc : SUCCESS in 3m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67abebbcd51c45f7abaf162c18af21a1 : SUCCESS in 3m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c233fa94f334dc9bd960f418874f840 : SUCCESS in 10m 59s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/5693a28723514dc19382b3f6b45910cc : SUCCESS in 11m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8e89c027fc64e9483fc4a87622bd2a3 : SUCCESS in 8m 49s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b3177dfc8c34d96a38abf5716d2df8e : SUCCESS in 8m 35s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a07906fbf3204cb78d01b4f73ef831ff : SUCCESS in 7m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0eaeb4b3be4b432d9f88d352facf3e6c : FAILURE in 31m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d320e6950264c97a6b6385c202e8b79 : FAILURE in 38m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae6fd2eba8ee4654975d41661ac3472e : FAILURE in 39m 45s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80245456303948dca4a3bcac79e59e5f : FAILURE in 37m 02s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8011cd0b5850413e9009e3a86124df84 : SUCCESS in 23m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd3dee8367564ebe9b4492be09523b36 : SUCCESS in 1h 02m 45s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/09ffe0464a394fd793ff6314ac8f13ee : SUCCESS in 1h 09m 30s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8d2383c155940939357a182903f2ff4 : SUCCESS in 1h 06m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7c279e87d89440708b61a10a23f967cf : SUCCESS in 59m 00s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d99ad932238f42ebb90efcf2fa75162f : SUCCESS in 19m 15s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/67e18604715e4ad48f26f89a282f45d3 : SUCCESS in 53m 26s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe1b00b29654454c9762bd2d2f41b706 : SUCCESS in 1h 05m 33s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd61c1a40e6c454483eeb44f0c244ee5 : SUCCESS in 1h 01m 10s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f59054e8aa7d42b0bbaf3a8fd37f1a48 : SUCCESS in 54m 19s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/20cfc50239ec4208afd764aed141fb5a : SUCCESS in 1h 35m 34s","accounts_in_message":[],"_revision_number":4},{"id":"370387495e3f2125c908acf92307e20595c07be2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-04-11 14:11:25.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"60c961f060a1cdb39959dba21ac7b5bd5ee8f914","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-04-11 18:38:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e84576f28b8c0eb23d762748bf4323da25a4b564","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 18:39:17.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Change 834120 in project openstack/oslo.log does not share a change queue with 834274 in project openstack/puppet-oslo","accounts_in_message":[],"_revision_number":4},{"id":"d3e23ddd55e38a5c5397618cd40bf154ea4fa2f8","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-04-11 18:39:34.000000000","message":"Removed Workflow+1 by Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"253d397b792d3261598a7f24c41ffc31e6f578e2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-04-21 07:53:27.000000000","message":"Abandoned\n\nwe decided to go with https://review.opendev.org/c/openstack/oslo.log/+/838190\nso no puppet changes needed","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fdb0455b8f4f248a7936c5cccaea428bf7937acf","revisions":{"b20e12f9a00896c33a6c5ebd837835c64314ebb4":{"kind":"REWORK","_number":1,"created":"2022-03-18 11:03:42.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/74/834274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/74/834274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/1"}}},"commit":{"parents":[{"commit":"95544711d4751a6dee6fee25a2695604094d31ca","subject":"cache: Add inet6 prefix for only memcached backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/95544711d4751a6dee6fee25a2695604094d31ca"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2022-03-18 11:02:55.000000000","tz":60},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2022-03-18 11:02:55.000000000","tz":60},"subject":"Add param to log Global Request IDs","message":"Add param to log Global Request IDs\n\nDepends-On: https://review.opendev.org/834120\n\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\nChange-Id: I8c34677059c2989f7700f27443c91c807d76160c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b20e12f9a00896c33a6c5ebd837835c64314ebb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b20e12f9a00896c33a6c5ebd837835c64314ebb4"}]},"branch":"refs/heads/master"},"9c9fb5146f9d79a444640f423349a28844d85811":{"kind":"REWORK","_number":2,"created":"2022-04-08 14:14:40.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/74/834274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/74/834274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/2"}}},"commit":{"parents":[{"commit":"95544711d4751a6dee6fee25a2695604094d31ca","subject":"cache: Add inet6 prefix for only memcached backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/95544711d4751a6dee6fee25a2695604094d31ca"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2022-03-18 11:02:55.000000000","tz":60},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2022-04-08 14:13:51.000000000","tz":120},"subject":"Add param to log extra context values","message":"Add param to log extra context values\n\nThis can be useful to log Global Request IDs\nalongside local Request IDs for OpenStack services.\n\nDepends-On: https://review.opendev.org/834120\n\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\nChange-Id: I8c34677059c2989f7700f27443c91c807d76160c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9c9fb5146f9d79a444640f423349a28844d85811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9c9fb5146f9d79a444640f423349a28844d85811"}]},"branch":"refs/heads/master"},"5a664634535a04c1442c51cc7565dbe61cbd5f12":{"kind":"REWORK","_number":3,"created":"2022-04-08 14:23:46.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/74/834274/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/74/834274/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/3"}}},"commit":{"parents":[{"commit":"95544711d4751a6dee6fee25a2695604094d31ca","subject":"cache: Add inet6 prefix for only memcached backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/95544711d4751a6dee6fee25a2695604094d31ca"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2022-03-18 11:02:55.000000000","tz":60},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2022-04-08 14:23:41.000000000","tz":120},"subject":"Add param to log extra context values","message":"Add param to log extra context values\n\nThis can be useful to log Global Request IDs\nalongside local Request IDs for OpenStack services.\n\nDepends-On: https://review.opendev.org/834120\n\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\nChange-Id: I8c34677059c2989f7700f27443c91c807d76160c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/5a664634535a04c1442c51cc7565dbe61cbd5f12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/5a664634535a04c1442c51cc7565dbe61cbd5f12"}]},"branch":"refs/heads/master"},"fdb0455b8f4f248a7936c5cccaea428bf7937acf":{"kind":"REWORK","_number":4,"created":"2022-04-08 14:27:38.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/74/834274/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/74/834274/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/74/834274/4"}}},"commit":{"parents":[{"commit":"95544711d4751a6dee6fee25a2695604094d31ca","subject":"cache: Add inet6 prefix for only memcached backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/95544711d4751a6dee6fee25a2695604094d31ca"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2022-03-18 11:02:55.000000000","tz":60},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2022-04-08 14:27:33.000000000","tz":120},"subject":"Add param to log extra context values","message":"Add param to log extra context values\n\nThis can be useful to log Global Request IDs\nalongside local Request IDs for OpenStack services.\n\nDepends-On: https://review.opendev.org/834120\n\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\nChange-Id: I8c34677059c2989f7700f27443c91c807d76160c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/fdb0455b8f4f248a7936c5cccaea428bf7937acf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/fdb0455b8f4f248a7936c5cccaea428bf7937acf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
