)]}'
{"id":"openstack%2Fpuppet-oslo~834473","triplet_id":"openstack%2Fpuppet-oslo~stable%2Ftrain~I8ff8886b65a323a65dd0ff9d01066c82a4116f8d","project":"openstack/puppet-oslo","branch":"stable/train","topic":"trivial-stable/ussuri-stable/train","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-31 06:23:03.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I8ff8886b65a323a65dd0ff9d01066c82a4116f8d","subject":"Remove release note file in wrong path","status":"MERGED","created":"2022-03-21 15:12:47.000000000","updated":"2022-03-31 08:42:19.000000000","submitted":"2022-03-31 08:42:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834473-trivial-stable/ussuri-stable/train","cherry_pick_of_change":746739,"cherry_pick_of_patch_set":1,"meta_rev_id":"f7931afbddbb047df344b4b91a83d89e9200dad8","_number":834473,"virtual_id_number":834473,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-31 08:42:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-03-21 19:10:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-03-28 20:24:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-03-28 20:24:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-21 16:25:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-21 19:10:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-03-28 20:24:31.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"080bd9ac38d62d97311a808063e8de3ea6915117","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-21 15:12:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"407a0f52c4720ecd19ebc594ca603b77753003d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 16:25:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48f1425d85fd427d92e74773a5ed4c51 : SUCCESS in 6m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3a36f3c9340410eb81f0159cd0fca33 : SUCCESS in 5m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0414dff9ac24463e96a041d4ef9765e8 : SUCCESS in 6m 59s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/498b0ebfd56e42dd9e50cdca07da12e9 : SUCCESS in 20m 26s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0a2b27bfa954bf8bf6b7482064f71df : SUCCESS in 14m 51s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/99e5107282ac466bbd7725380a13458e : SUCCESS in 12m 57s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b9e8b1a3f8e4e8d8a202f07488b98e6 : SUCCESS in 13m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a175ce5e6304dddae6757c1dddaa55c : SUCCESS in 10m 43s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/6ff77c16b3e449aaa2fb1a2d92e8c1aa : SUCCESS in 55m 42s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e1d18462bf95475ea2802609f3cca2e2 : SUCCESS in 1h 05m 37s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/01f5aef6b6fc44a1a4298915c9c1e6d8 : SUCCESS in 1h 04m 27s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/6786e3246c4c4eee948f6a1bea1659e5 : SUCCESS in 50m 37s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/cce6b36ec00145ffb0b862145bc559b4 : SUCCESS in 43m 06s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bcd539d7aaf4a7fb1cd3a0f94f32314 : SUCCESS in 44m 05s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5dd606b7028345d4ba91ca4e508989d0 : SUCCESS in 39m 22s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ec92fd87e3024865a7001f2b42d70486 : SUCCESS in 36m 31s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/91a19ba17d0b424bad0d9ec3e8495ebd : SUCCESS in 52m 39s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/037e5f789a0a4baea73f743ddc31d542 : SUCCESS in 1h 03m 24s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/21e2c32964644a86afa23b823e0caa66 : SUCCESS in 1h 00m 03s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/802459aeb5824df6ac464f9b89a2db57 : SUCCESS in 45m 10s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/380a5b5cf8b047cb875c0a7da7873a66 : SUCCESS in 11m 05s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ffc78fedaa44461f9b2a909b36c9ed49 : FAILURE in 9m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7cfd654c8da5ba7097398866bde07792467c59b7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-21 19:10:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"65eb8f1700bf315b19328bd998fe0bf7d9935f9b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-28 20:24:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81bf0b98f5a67d70602b918bdde0d2d2749c18df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 22:12:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"729e7da59fe119fdff57dc1760ad5a7561959870","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 23:34:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7f09e9376fc4fe59a3bb7edce2c4613 : SUCCESS in 6m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2cdbd686a1c4a5386c271e54d173210 : SUCCESS in 5m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40dab067af084b33a0925d73e14c653b : SUCCESS in 6m 40s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ad8927afaf4e45b297a9b71e7ee52335 : SUCCESS in 21m 11s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/214e500140454afab50ce001b629427e : SUCCESS in 14m 03s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/c9290f61f3254b779a9ce2cc9fa15d4e : SUCCESS in 58m 31s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ceb124c9a7054af3a9097b5c9d5bc920 : FAILURE in 1h 13m 27s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9c15e6818ff54f4caa700ef13b75a174 : SUCCESS in 1h 09m 34s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/cd4b55caba4f469d91de56b641b886e8 : SUCCESS in 53m 21s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ba735cf9dc064eb2a41f03992712866f : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":1},{"id":"28d54216acc5cdc80f0cc91c061811a6ea15d334","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-30 00:29:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"57f2813b9b679010e0a3f3d84f8358591b38f2e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 01:48:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0767995c0fd84e3da83386ad0f310d64 : SUCCESS in 6m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e9b694ee0ac4a62ac77d6e9f2ee6335 : SUCCESS in 5m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70b0650cb5d74c8d9750c5fae813a893 : SUCCESS in 6m 21s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0eb5506738ad4b62924518a1da67228b : SUCCESS in 19m 54s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/02369fb77c374c688a3bbc095140b29a : SUCCESS in 15m 44s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/f562c5b1d07b4535a00174c4a37f8294 : SUCCESS in 15m 28s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d13f326915b42a0b9712a0c258e3dd0 : SUCCESS in 12m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e5d1f6bd8104711b91c63996238fafb : SUCCESS in 10m 26s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/425b850c36f146d0b75f3847656dd7c5 : SUCCESS in 1h 07m 10s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/05e48ba4af7a486f840b7de1514ffd00 : SUCCESS in 1h 06m 38s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f1fc44c1ac2648548b50b74d7b69a9c9 : SUCCESS in 1h 05m 10s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/19030059178146ac8e99c2c59b687936 : SUCCESS in 53m 30s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3b13963c982a48cbb6601c5dc28cd8c9 : SUCCESS in 46m 57s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6278dd1302104d7cb91b585e10e81b1a : SUCCESS in 59m 33s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20adecca9fe1423bbda5f89d41366ec8 : SUCCESS in 38m 17s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6ad292018556404b8373c82f4fbe4efd : SUCCESS in 36m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ed32974970e414b877047b851e126c3 : SUCCESS in 1h 13m 06s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4a0e7a4954343f88cbad79fade560de : FAILURE in 1h 12m 08s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ab6e60dc80740dd9cf57991760a607e : SUCCESS in 59m 45s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/20d54d1a64c74de79a060a267294b64f : SUCCESS in 47m 59s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/82e4b2da05754a019828551444a044c0 : SUCCESS in 11m 01s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d496032abfe447fa4fb7849a63f8254 : FAILURE in 10m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27d2e97700d6d175145fda9eed81912c043bb56a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-30 05:17:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e327be91cbb39c9c069b89e7a82b7288d69c111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 06:24:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb03a0b4b8e54e668112d212ec1f7446 : SUCCESS in 5m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b48f1989c4644fd88c5013be5539ed5f : SUCCESS in 4m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/061bb988eee440c2854ddb7db927522d : SUCCESS in 5m 27s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/dd2a3f84f11e4f779f0b686af9342618 : SUCCESS in 12m 03s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/2613bf7c4ab24a76bc755ca8faef58b3 : SUCCESS in 12m 40s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5960cccbd6e4acbbeed5318482cae1e : SUCCESS in 12m 07s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13d9ce50ac4045f18ac4c5e04ed60c1f : SUCCESS in 9m 41s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3bec3175e9b432ca22251463e091c2f : SUCCESS in 9m 21s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/07ffbe1787f5405baa5aec0eb363c8d1 : SUCCESS in 52m 25s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0ec0d39d70b0463fac7c5630d97ccf36 : SUCCESS in 1h 05m 12s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6f7b657078d14cc0b46caf57b40f5c18 : SUCCESS in 59m 06s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/a224e0147f08476dbd50610bfbc344fb : SUCCESS in 44m 20s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/95421d5be02f45b084cecef77018c7e3 : SUCCESS in 41m 47s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5322bd8ce8d441f59bc6fe1e05aa1e98 : SUCCESS in 44m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d6b5b980e2e41f6b729243185a52ca1 : SUCCESS in 41m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/263b5473e3964b07a7d87b55c9453324 : SUCCESS in 38m 06s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/67cc7199fe8541dab95e912c44db51fd : SUCCESS in 52m 57s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/284e5642b9a24edcac85b6fcabfdb44b : SUCCESS in 58m 34s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b3eac70bfaf04da49265a8018bc8ea54 : SUCCESS in 1h 00m 58s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7c3d44c823834619a5fb262220fb8d5a : SUCCESS in 48m 19s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/151279e87afa4be6a7e21451616fdc68 : SUCCESS in 10m 21s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03b5b80a5136465eae45621d0d5e4dfc : FAILURE in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a62168a981831a25b34b477cf8801811bc152a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 06:25:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a9e55a4c23bf04c6685405f69fc639f69e1fc140","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 07:59:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/131b0e2659824de0a7c2c817781ecc34 : SUCCESS in 6m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3972039d68fa4c579eba9096b1c00234 : SUCCESS in 5m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4814b01d4f524c9da690cb8f0e8b8072 : SUCCESS in 6m 11s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a121a3fc22934c0db8603fe6ec4bfa70 : SUCCESS in 11m 05s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc423f5f357f4784842b02a6d87ae294 : SUCCESS in 11m 59s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a60cb4dace854dac82f5eb8795c189db : SUCCESS in 53m 23s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/88d7a95f4b564e87881f54dd1186d9da : FAILURE in 1h 29m 20s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a1ff792c4d23421bacc5ca03b10ac1ee : SUCCESS in 58m 31s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/ab1c823b09884acfb7978173c24fc802 : SUCCESS in 54m 55s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5781130eae034911af78f5243f92d344 : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2727d896662e56d0b3e79a14aaf008f5a56887dd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-31 06:23:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c841d706c46235bb0d5b495dee481a77e1b6f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 07:35:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1411b467ac3e4f73a1ddea5907e9ff3a : SUCCESS in 4m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a37ee35518aa4238bfd68521a321b1df : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9648a6d01ec74012a57a53d01d2c7542 : SUCCESS in 3m 57s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/46aa6518327a427880393360fd34e79f : SUCCESS in 9m 49s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb80b15a692a40cea85138383350e97e : SUCCESS in 10m 29s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4542ba677ed491f9be25b257653945a : SUCCESS in 10m 59s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84038bfe9a4f42569e5dbbda41ebf123 : SUCCESS in 10m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/142afef728e34255ba1f53f9c7972c2b : SUCCESS in 8m 30s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/e3a5a6e24b1b42e0b78fb367e54eaa39 : SUCCESS in 1h 00m 03s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0057bacdd1724cdebc4eac27425ddf89 : SUCCESS in 1h 03m 21s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6894f26479ea4503971b63985caa40a7 : SUCCESS in 59m 47s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/411f1ec5f9ed4b7e8504689ab54c0ea9 : SUCCESS in 48m 22s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4c54c6f5fa2245d5a97ed1f06e1c813b : SUCCESS in 41m 16s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50aeb2d2a8a74826b05849597012492a : SUCCESS in 52m 29s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8deeff3573c5465ab71a57a9fac63b27 : SUCCESS in 41m 31s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f9a19e0a770f4a618f717df9f166dfc8 : SUCCESS in 36m 54s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f14687e018ab488ca88335d247b92558 : SUCCESS in 52m 14s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/87cf0760213a4e3fa0fc484f004783fb : SUCCESS in 1h 00m 30s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e329808d4b184317b922ea4775a3421a : SUCCESS in 59m 49s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b918df60435249c2a61a3148dd80ae85 : SUCCESS in 45m 18s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7095cf732d304a349cc8e250a60eed42 : SUCCESS in 8m 56s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e33db646ca2d464aa58de26d642177fe : FAILURE in 7m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8624d30c07b85697b085698615f79acbc00e27b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 07:36:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7459bbc0e23ac33f31921bf50dfb5c6b6b5804fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 08:42:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9728f21f97424b839431f8e1e2b06a83 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/439ad193e1834901bc14d6cefb0def61 : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd4d2d82578945eaabcf8e0710da3298 : SUCCESS in 4m 10s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a88e0cc62ceb42c9b2bdb97c2ba72661 : SUCCESS in 17m 08s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7b87646234a482881d6c77f69fab1cc : SUCCESS in 10m 48s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/797d2c1dd75a4ae6859e52d0f8839f8d : SUCCESS in 1h 02m 01s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8fc8ab5ff583401fa3ef46466c85dbf4 : SUCCESS in 1h 02m 19s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/24dea061173b4199b97af5aa7f2d9e35 : SUCCESS in 1h 02m 27s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/229f0f4741ed4270ad4ece2a5753c0d7 : SUCCESS in 50m 17s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6baf92e2adf14c099d36d50f54c7b179 : SUCCESS in 10m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f7931afbddbb047df344b4b91a83d89e9200dad8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 08:42:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"052053d1c60b1a83e2dff75e9c5174b0ffe3c3f3","revisions":{"052053d1c60b1a83e2dff75e9c5174b0ffe3c3f3":{"kind":"REWORK","_number":1,"created":"2022-03-21 15:12:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/834473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/73/834473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/73/834473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/73/834473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/73/834473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/73/834473/1"}}},"commit":{"parents":[{"commit":"b4f87e89cd5376343cb35c00366629abfa28325a","subject":"Fix the wrong default value of idle_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b4f87e89cd5376343cb35c00366629abfa28325a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-14 00:20:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-21 15:12:47.000000000","tz":0},"subject":"Remove release note file in wrong path","message":"Remove release note file in wrong path\n\nThis patch removes the release note file which was added into wrong\npath in an old commit[1].\nThe removed relesae note file is not added into releasenote directory\nto avoid confusion by a note associated with wrong release.\n\n[1] 0f617ce27a99260ee03627b1fbde461a687718dc\n\nChange-Id: I8ff8886b65a323a65dd0ff9d01066c82a4116f8d\n(cherry picked from commit 064edacff36d949ebdd304bd404a5fdcff1f8981)\n(cherry picked from commit 5fa0fca1bee859ff31357c7f0df50ee4d9b73145)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/052053d1c60b1a83e2dff75e9c5174b0ffe3c3f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/052053d1c60b1a83e2dff75e9c5174b0ffe3c3f3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
