)]}'
{"id":"openstack%2Fpuppet-oslo~836243","triplet_id":"openstack%2Fpuppet-oslo~master~I8c4358128438700430645df132882ec46a8885b9","project":"openstack/puppet-oslo","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-04-05 15:46:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8c4358128438700430645df132882ec46a8885b9","subject":"log: Accept array for default_log_levels","status":"MERGED","created":"2022-04-03 14:14:24.000000000","updated":"2022-04-05 15:46:23.000000000","submitted":"2022-04-05 15:46:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836243","meta_rev_id":"eaccc0166eff973b6b659f3f622ae883a255d5ec","_number":836243,"virtual_id_number":836243,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-05 15:46:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-04-04 19:59:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-04-05 14:02:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-04-05 14:02:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-03 15:16:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-04 19:59:02.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-04-05 14:02:52.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"24399e8956846e37b30ca7a0a0ebfefaa89439aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-03 14:14:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f35fa4b819bf83fcc7504e7c8d3460c29319923a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 15:16:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eca047414e334530ad524dbf61f5d27f : SUCCESS in 2m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d134f45655d47cfadbad4dda1f9826f : FAILURE in 2m 52s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8c3955b3f10474b95ac8c174627f92b : FAILURE in 2m 45s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/161b875958e04862868d0d805a30b795 : FAILURE in 10m 44s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/0cbc034bbbf2435fbd6e8e3e7f59a3d7 : FAILURE in 9m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2039a24ebdcf433d8ca1393e0314a6af : FAILURE in 7m 12s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abd530a18fdf46189401f105dec073a4 : FAILURE in 6m 39s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a06346847e24376a10bf0f2e8579b59 : FAILURE in 6m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2de4fe162b5d4651be28ff5a6fd50a9a : FAILURE in 7m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82d8f8a0488d475b8f655c5453e8c751 : FAILURE in 7m 05s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3630468034a400396c3e02caaf039b5 : FAILURE in 7m 39s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f09cc849c61452f9eedc4d3a9452881 : FAILURE in 7m 33s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/773307da78a24051a1a5cc657af0e2c4 : FAILURE in 10m 39s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/528bdee204c14e2abe2e70b7df96f86e : FAILURE in 11m 36s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5c30fee64c194de59cfee631e7fc5809 : FAILURE in 11m 26s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e8378e65690945d2a68264bff40eef52 : FAILURE in 10m 46s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5297a273731d47a2aa157623c355c3a9 : FAILURE in 10m 13s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/06647caedf9c4848ba4c6fe084b9b0b7 : FAILURE in 9m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/262bc240ea4e4ec9a7ee5455cbbccc65 : FAILURE in 9m 41s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/63a23e32dbf3457681ed6261bb8805e8 : FAILURE in 9m 36s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/821203ac75b046d5b65c73b22a2a3b07 : FAILURE in 10m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c8a0d5e5d6946dbbfed079b62019c26 : FAILURE in 9m 43s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/536d0103ce9c4fbb8b5b248b15dd7a52 : FAILURE in 51m 19s","accounts_in_message":[],"_revision_number":1},{"id":"aa216a42b5f4d78e1e2911a144ab28bf41b78dc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-03 15:52:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"16962f1c53f1fa7970bbc54c5ced63e78f4de044","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 17:36:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f342edeb00de488dafbe2624ecc0dd56 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/579df27bc1124d61ad27867c43ee79c3 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5924008d3adb40e2b9b9cce63a4600b8 : SUCCESS in 3m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d5f6ba08a8b4f8499e047e5766c47b4 : FAILURE in 10m 56s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/3dad1948b17745aebf223b13a8d83715 : FAILURE in 10m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fadbe31bf3c4b48b742834485e852cb : FAILURE in 6m 45s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41e698d074e34bec94e0b0b85d968cee : FAILURE in 8m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e8f48c9ae7343d59efa555bf90d380f : FAILURE in 6m 48s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7959c18db814efeac264b5036d659bf : FAILURE in 30m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e82e229e733489e8a05370a908e2a76 : FAILURE in 40m 12s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dd75f11d58540b9bef60950d8f1b271 : FAILURE in 38m 52s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a52263e53d7f4f50a388be38ab2beabd : FAILURE in 30m 56s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe19065070084b6387620164eb5459a5 : SUCCESS in 21m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e011cde79dc144d0b376efdcdc257231 : SUCCESS in 48m 33s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3857c5b991b43b0a3da671d9a8fcecb : SUCCESS in 1h 06m 08s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/99f86b3752a04571a0cde972ea16055a : SUCCESS in 1h 00m 42s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/831586cc72e746e29b73a8907bb2dff7 : SUCCESS in 48m 29s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9eb99685a9f64a36947705b4593a8c79 : SUCCESS in 16m 04s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6efdc173988448895d5e5c70d98a60a : SUCCESS in 45m 37s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/82c85034c6294c919127ed35d18fe850 : SUCCESS in 54m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c6897e996924fd1a5ab4d1f77286703 : SUCCESS in 57m 21s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b0859bc99f4452f82a063d37fbe4297 : SUCCESS in 41m 46s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/db8d996f5b6349e8ac7f4da1fb6f8495 : SUCCESS in 1h 25m 38s","accounts_in_message":[],"_revision_number":2},{"id":"fd1de1fef8554eba808f8e2e83d30ab14ccace44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-04 04:55:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"018aa8127370ab7df2897985ebe1f41f27759a53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 06:48:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d54a599107d491783cd237f37a9b025 : SUCCESS in 2m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e30f53a462364cab8fceab7c7097a4ab : SUCCESS in 2m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f14dd723dc04ace8bbd35e51694f3fd : SUCCESS in 2m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a1164b6f4fb4bf98df3bab94da2fae2 : SUCCESS in 9m 30s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/710e5b7afeba4fc6a4d821f1e9c827f2 : SUCCESS in 9m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c6f0f9250f940ca875756c91f1154aa : SUCCESS in 7m 13s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3884abbbcdb42d2a43b35900edff798 : SUCCESS in 6m 48s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8b32737f20e4431b5f9a6ff1b9512d4 : SUCCESS in 7m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fb88b621af049d0988bad86ac47b18a : FAILURE in 28m 43s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f2c29d291b14bb7a2cdf921ff9773bb : FAILURE in 32m 26s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6700dc2773244a38a59d89af3a0844e9 : FAILURE in 32m 58s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60b76b2435da4ff19beda94b8d95eca5 : FAILURE in 37m 10s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f1649d9ff5d47d281f12a911ed4b3d5 : SUCCESS in 18m 53s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed3a72f6696849a1a913a267c8495e71 : SUCCESS in 56m 46s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef2a9d10d59749418127ba5c3c63b815 : SUCCESS in 1h 12m 26s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6c09520722949de9ba863f25d8a419d : SUCCESS in 59m 28s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa5d9e3917534a3d89b0fd5d50e8d71a : SUCCESS in 45m 57s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/35ad9c83d247498eb0100a010c014966 : SUCCESS in 16m 17s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1d9ebc8976842f2a7f4b9393c0834c9 : SUCCESS in 49m 10s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b481142562ed4083b90d62cf530560c7 : SUCCESS in 56m 38s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bebb5e555da48039307815bbe97a597 : SUCCESS in 54m 51s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0de04f83122d4a7291b132d7ddff39d6 : SUCCESS in 48m 19s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a59cfc9f79ca4fb3918b2ad596881f1e : SUCCESS in 1h 17m 14s","accounts_in_message":[],"_revision_number":3},{"id":"f0e37138679afa541d444e97c8ca0bfaab2d8a32","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-04-04 19:59:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bb97a7e225cf92ebcc702824bbc55bc441f217b2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-04-05 14:02:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7ed4c879ea29fb3d9d1c8e5c5870f01d7adea374","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 14:05:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"631d8c04882fe2bafb42e440d8eab31b0f038312","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 15:46:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9619e7b72624bae97cdfcc00fd1042a : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25d6cc6607624d919219fb94c14bab45 : SUCCESS in 4m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5514afcd5bed425497f2d9477fc8f0ac : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a1effc8015994bf9a9c2de3e741fc3e2 : SUCCESS in 10m 42s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb946f32159e44e0b4f35aee1c972dbb : SUCCESS in 11m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43f4a3a753f44354ab9c4981c198537e : SUCCESS in 10m 56s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4322e615148f4076aa0ceaf9f53178d5 : SUCCESS in 11m 46s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/178a3e37a1de42f196363ac38debad31 : SUCCESS in 1h 10m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c8007556eaae4116a3949396dcdc4767 : SUCCESS in 1h 17m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bfcde945b47a466d85a16a3331a54c56 : SUCCESS in 1h 08m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5dd8136a908c4291bfff20fb3dfd023a : SUCCESS in 1h 01m 03s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a44ff25c013945a39624d14299297958 : SUCCESS in 1h 36m 39s","accounts_in_message":[],"_revision_number":3},{"id":"eaccc0166eff973b6b659f3f622ae883a255d5ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 15:46:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"efd11469933af2b0aadc54546ce611b92c66391e","revisions":{"640069654bf39b18af054a0375d765de4ffd8e27":{"kind":"REWORK","_number":1,"created":"2022-04-03 14:14:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/836243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/43/836243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/43/836243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/43/836243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/43/836243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/43/836243/1"}}},"commit":{"parents":[{"commit":"fe6612cb6c40b5e02425caadb7aa93a5c0a23f29","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/fe6612cb6c40b5e02425caadb7aa93a5c0a23f29"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 14:13:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 14:14:14.000000000","tz":540},"subject":"log: Accept array for default_log_levels","message":"log: Accept array for default_log_levels\n\nCurrently the default_log_levels parameter accepts only hash or string.\nThis change adds support for an array value, so that users can use\nan array of \u0027logger\u003dLEVEL\u0027 strings.\n\nChange-Id: I8c4358128438700430645df132882ec46a8885b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/640069654bf39b18af054a0375d765de4ffd8e27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/640069654bf39b18af054a0375d765de4ffd8e27"}]},"branch":"refs/heads/master"},"e036747e4b03f34d56b5d2601234c58804987542":{"kind":"REWORK","_number":2,"created":"2022-04-03 15:52:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/836243/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/43/836243/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/43/836243/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/43/836243/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/43/836243/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/43/836243/2"}}},"commit":{"parents":[{"commit":"fe6612cb6c40b5e02425caadb7aa93a5c0a23f29","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/fe6612cb6c40b5e02425caadb7aa93a5c0a23f29"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 14:13:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 15:52:57.000000000","tz":0},"subject":"log: Accept array for default_log_levels","message":"log: Accept array for default_log_levels\n\nCurrently the default_log_levels parameter accepts only hash or string.\nThis change adds support for an array value, so that users can use\nan array of \u0027logger\u003dLEVEL\u0027 strings.\n\nChange-Id: I8c4358128438700430645df132882ec46a8885b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/e036747e4b03f34d56b5d2601234c58804987542"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/e036747e4b03f34d56b5d2601234c58804987542"}]},"branch":"refs/heads/master"},"efd11469933af2b0aadc54546ce611b92c66391e":{"kind":"REWORK","_number":3,"created":"2022-04-04 04:55:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/836243/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/43/836243/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/43/836243/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/43/836243/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/43/836243/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/43/836243/3"}}},"commit":{"parents":[{"commit":"fe6612cb6c40b5e02425caadb7aa93a5c0a23f29","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/fe6612cb6c40b5e02425caadb7aa93a5c0a23f29"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 14:13:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-04 04:55:01.000000000","tz":540},"subject":"log: Accept array for default_log_levels","message":"log: Accept array for default_log_levels\n\nCurrently the default_log_levels parameter accepts only hash or string.\nThis change adds support for an array value, so that users can use\nan array of \u0027logger\u003dLEVEL\u0027 strings.\n\nChange-Id: I8c4358128438700430645df132882ec46a8885b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/efd11469933af2b0aadc54546ce611b92c66391e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/efd11469933af2b0aadc54546ce611b92c66391e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
