)]}'
{"id":"openstack%2Fpuppet-oslo~836245","triplet_id":"openstack%2Fpuppet-oslo~master~If524155bf2d0dda964c6b451d7b26f36481514f5","project":"openstack/puppet-oslo","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-04-05 15:49:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If524155bf2d0dda964c6b451d7b26f36481514f5","subject":"Split oslo::service defined type","status":"MERGED","created":"2022-04-03 14:39:59.000000000","updated":"2022-04-05 15:51:22.000000000","submitted":"2022-04-05 15:49:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836245","meta_rev_id":"33b19aa9b094a7b8e844ba1f70bd8d4053078474","_number":836245,"virtual_id_number":836245,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-05 15:49:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-04-04 19:55:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-04-05 14:02:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-04-05 14:02:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-03 16:12:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-04 19:55:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-04-05 14:02:58.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"bd3aecd4329a206f8e6eed27933f599454fb7ac2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-03 14:39:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60095d3bba97449b0aa1a34b278bfbe734fec23d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 16:12:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84ed1df14f274b48bf3ef36e9d279f5f : FAILURE in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29af8fa0db464cf99e3181bdb5c00895 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30a90ab790a54035b98a97be202bc0fe : SUCCESS in 3m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8711cf521f743da838332837c2c2384 : SUCCESS in 9m 52s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2da079a5cc34df19ab047362f9fbfc2 : SUCCESS in 9m 59s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ccd81d84ade4f559dbbcd0ce94a6756 : SUCCESS in 7m 13s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4f08980ccfc49be9fb37e79db8c7c62 : SUCCESS in 7m 25s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1cb9fa88e164d3b88ed87d936832676 : SUCCESS in 7m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfb53bd2cfce45ffb37d421915a7600c : FAILURE in 32m 05s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4540c1930a6e4e6486f6d5781423608f : FAILURE in 33m 13s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3475098a807841b39b3cd8848b67737f : FAILURE in 25m 55s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4dfeccffdbf4955816fc55f4df1306b : FAILURE in 34m 33s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4ec2d49ca404a7486c32257c20fc7f7 : SUCCESS in 18m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/041e281037374742825d507cf3e3bf92 : SUCCESS in 1h 01m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ae68523cc5849f5af8bbb612191eeb8 : SUCCESS in 1h 08m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5edeecc8a07f4c859b3bd5c317c2fe61 : SUCCESS in 1h 01m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/307c7f4d63a040da8de5ff52bef9f4f0 : SUCCESS in 58m 34s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cda4b41e57854ef0aed714783f4e43f8 : SUCCESS in 20m 10s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9dde057e02ca47c89764fafaee777641 : SUCCESS in 49m 39s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fff03b2434c84205b3363f11abe4f610 : SUCCESS in 1h 01m 17s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca8e76bc4fe1403b95e3d5437f235c74 : SUCCESS in 55m 58s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a40480ae795a47dc9a7aef9530aa270f : SUCCESS in 49m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef388424d97b436fb503cb332dad559c : SUCCESS in 3m 40s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4c39893e72bc49edbf6ba264051ba021 : SUCCESS in 1h 29m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a9b918924c8e494eec5aeb93f5e309e132e27950","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-03 23:48:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"200d807d02ebef93349ad88f22ca81c0314937de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 01:00:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b27eae4b2d3f498a8a1da575c2c52d15 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b82c39fb576448d88f78cc012c8a9c98 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cade1c7f60af4d4fac734cc99b693604 : SUCCESS in 2m 45s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d490349a9e11403d99ca4f80364a3239 : SUCCESS in 10m 18s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/143367e88b694dd784108f35de0d25ca : SUCCESS in 9m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db7a37be458f44798056ad8212b36c6e : SUCCESS in 6m 52s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/596f318c430041f89b2be82fc1f31703 : SUCCESS in 6m 56s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8469df08e52544a589a735203402eed9 : SUCCESS in 7m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf575d23ca8b4729a7536df2070a6177 : FAILURE in 27m 37s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/daaf5a762960408ca89f8c808b2021d3 : FAILURE in 32m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14b8cde4dc424ab09cc4f352acb27d99 : FAILURE in 28m 36s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fce0be85e63f4006b5b00b8fac904fc9 : FAILURE in 34m 40s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9c75499157040f990aa09f40b29b800 : SUCCESS in 18m 20s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9015462d42a24f559660b89ae0742c20 : SUCCESS in 48m 15s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a78bd908283f44bbb230bdb7474e7a83 : SUCCESS in 1h 03m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c10f48aa09cc4984917cf1d7e206bd33 : SUCCESS in 58m 45s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/35c654c8bf164495b7dedd6c795a8c66 : SUCCESS in 43m 15s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a894ba084f64a259157b901b5f50d6c : SUCCESS in 20m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2136396e73042b69263233bfb85cdfe : SUCCESS in 38m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/be4577d50d4945789ec266e5ecbb39fa : SUCCESS in 51m 02s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d34ba5a310b4a1f91f2b102ee9e05a8 : SUCCESS in 47m 54s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a04930759bfc4e4fa8d05f9b908a5cf0 : SUCCESS in 47m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7c2eecceb38436b9429a976361bb430 : SUCCESS in 3m 09s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e250fdeb687a4b0da983fd1f3a9a1903 : SUCCESS in 1h 09m 47s","accounts_in_message":[],"_revision_number":2},{"id":"65870ab2bccfa8e6325647081a9cba7a384313fb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-04-04 19:55:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ef38ea09c0d0059b31e4af794d07dc147b4ffe98","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-04-05 14:02:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"db3987fd4bbf5ce75f2cea678869f55f4840ff61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 14:04:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1eb2e851e636869cff1e1c19c7ad2471d437019d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 15:49:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea660f877fae4d1f93513133c6455ea2 : SUCCESS in 4m 09s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9294569403d418ea356a25071960a28 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee25fc8691214d73a6655d713a74197e : SUCCESS in 3m 47s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2bb200c9a704473390845f2a146e9cc1 : SUCCESS in 12m 01s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/543e5d0e880a49a9a56342165fd357c6 : SUCCESS in 12m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b348a2c523d94b3584d24ee3a989edf7 : SUCCESS in 9m 22s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c0af12f5cf24b93a7f75ce41f320649 : SUCCESS in 11m 02s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ade684365cc4f9a81a6ac911c34b8ca : SUCCESS in 1h 04m 38s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca473cbdb06643e2a1df11cec913e5de : SUCCESS in 1h 16m 47s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/974b8df9e6b147f39987ebfb9c077889 : SUCCESS in 1h 08m 13s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/408644d4663f4366a6f8b7f3be73feb4 : SUCCESS in 1h 04m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3301a516f9046c1a6709d1513d71b01 : SUCCESS in 4m 25s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3be90ee196144953a3a60e48d1d16aa4 : SUCCESS in 1h 38m 46s","accounts_in_message":[],"_revision_number":2},{"id":"cd1974c6642cef0718cfde9ad181af536fb628b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 15:49:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"33b19aa9b094a7b8e844ba1f70bd8d4053078474","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 15:51:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27f0d41618fe4479be2dd59dbbb7e96d : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"19862b0f1879f41ef364d0fdac07df9415bdf2d3","revisions":{"dd67d759ca4479446be9c04644ebe7148b9d5fdf":{"kind":"REWORK","_number":1,"created":"2022-04-03 14:39:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/836245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/45/836245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/45/836245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/45/836245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/45/836245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/45/836245/1"}}},"commit":{"parents":[{"commit":"fe6612cb6c40b5e02425caadb7aa93a5c0a23f29","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/fe6612cb6c40b5e02425caadb7aa93a5c0a23f29"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 14:34:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 14:39:05.000000000","tz":540},"subject":"Split oslo::service defined type","message":"Split oslo::service defined type\n\nThe oslo::service defined type was added a long ago but has never been\nused by any other modules so far. The main blocker is that the type\nmanages not only common service parameters but also wsgi/ssl parameters\nwhich are specific to api services.\n\nThis change splits the defined type to smaller modules so that we can\nreuse the implementation.\n\nBecause the defined type is not used by any other modules at this\nmoment, and we don\u0027t expect direct usage of these resource type(*1),\nthis change is backword-incompatible.\n\n(*1)\nTechnically it can be used but cause multiple conflicts.\n\nChange-Id: If524155bf2d0dda964c6b451d7b26f36481514f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/dd67d759ca4479446be9c04644ebe7148b9d5fdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/dd67d759ca4479446be9c04644ebe7148b9d5fdf"}]},"branch":"refs/heads/master"},"19862b0f1879f41ef364d0fdac07df9415bdf2d3":{"kind":"REWORK","_number":2,"created":"2022-04-03 23:48:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/836245/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/45/836245/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/45/836245/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/45/836245/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/45/836245/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/45/836245/2"}}},"commit":{"parents":[{"commit":"fe6612cb6c40b5e02425caadb7aa93a5c0a23f29","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/fe6612cb6c40b5e02425caadb7aa93a5c0a23f29"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 14:34:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 23:47:54.000000000","tz":540},"subject":"Split oslo::service defined type","message":"Split oslo::service defined type\n\nThe oslo::service defined type was added a long ago but has never been\nused by any other modules so far. The main blocker is that the type\nmanages not only common service parameters but also wsgi/ssl parameters\nwhich are specific to api services.\n\nThis change splits the defined type to smaller modules so that we can\nreuse the implementation.\n\nBecause the defined type is not used by any other modules at this\nmoment, and we don\u0027t expect direct usage of these resource type(*1),\nthis change is backword-incompatible.\n\n(*1)\nTechnically it can be used but cause multiple conflicts.\n\nChange-Id: If524155bf2d0dda964c6b451d7b26f36481514f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/19862b0f1879f41ef364d0fdac07df9415bdf2d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/19862b0f1879f41ef364d0fdac07df9415bdf2d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
