)]}'
{"id":"openstack%2Fpuppet-oslo~836925","triplet_id":"openstack%2Fpuppet-oslo~stable%2Fussuri~I7deed1bca1ce2f715967e966d4b0cedc1f227027","project":"openstack/puppet-oslo","branch":"stable/ussuri","topic":"ussuri","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-04-11 15:52:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7deed1bca1ce2f715967e966d4b0cedc1f227027","subject":"Add socket keepalive options for the pymemcache backend","status":"MERGED","created":"2022-04-07 10:02:33.000000000","updated":"2022-04-11 15:54:43.000000000","submitted":"2022-04-11 15:52:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836925-ussuri","meta_rev_id":"68b611a4a824544d86ace10ad51b629ee65aaa37","_number":836925,"virtual_id_number":836925,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-11 15:52:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-04-07 11:25:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-04-11 14:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-04-11 14:11:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-07 10:02:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"CC"},{"updated":"2022-04-07 11:17:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-07 11:25:26.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-04-11 14:11:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"d69903d1cd7dfbe5e04a112011697fd80e3405c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-07 10:02:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49e929843d8e0ab6d766e91f66f5dee7898336c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 11:17:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebf3d15411f34118aa676ad8395e5fa1 : SUCCESS in 3m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f88521c648b8465ba2680edc8fbefa76 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cf97a488f9542db8965dd85c7186430 : SUCCESS in 2m 57s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/e11bf6d62c8748f49c2ee36d5bd2662b : SUCCESS in 12m 34s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/76032b0dc9914869848ed0a98cfd19c0 : SUCCESS in 12m 40s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f543ead81f74e70862ebc1869cf8ec4 : SUCCESS in 10m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d12b8a515f547f6ab51d77e1fd751d7 : SUCCESS in 7m 04s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/414ba013423e4d18bcd35ec9fac9448e : SUCCESS in 51m 31s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44bb19f254f94568ab3055812e701c8f : SUCCESS in 48m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7bd92073f6a41bc9fdf3324fa260bb1 : SUCCESS in 40m 18s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a597c5f4adff4df5be27758cd9560541 : SUCCESS in 35m 46s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a213a6d15c849bb8b6c1a86333486b0 : SUCCESS in 58m 16s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3eb53835976c4e83842b7de1515ae4fe : SUCCESS in 1h 08m 39s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e41c04f072fa447d90aed245de36fa46 : SUCCESS in 1h 08m 18s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e36c2f064c24e3f84da15269ee51576 : SUCCESS in 49m 56s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d7231070a80408082a9eeabde80f2f9 : SUCCESS in 8m 42s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad32093a7f5249feab479461c08ad985 : FAILURE in 5m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/413c66536501479aaabb9037b4158b63 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"410fdc8c25d6c6d210184dd3520201e52ed92f92","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-07 11:25:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ea418cab076cd958881183ca6cb08ba38aecc7c4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-04-11 14:11:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0c7a973469b1b6b1195d8c98b41c8954a457690","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 14:12:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25876a3abe8de053af64a5ba4915bc2819adfae7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 15:52:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96a0f6a690c340368ceeaa8721e21413 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8bf0f761e5d49bb8b3eb617692b0c1c : SUCCESS in 4m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6db896926123400b970ee167725187c3 : SUCCESS in 3m 33s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/5098d91b791446f08783550ea97ab593 : SUCCESS in 14m 22s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/9caaf89a24bd48dbacb738be903beecc : SUCCESS in 14m 50s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64e8a2c72f2b4b0c80ebecabb41d94ad : SUCCESS in 11m 20s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/075bd72876084792839898300fa79981 : SUCCESS in 1h 01m 18s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/01b493914d2946969c12ad13f6ee280a : SUCCESS in 1h 10m 21s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/186acbe0abe8451298cad83c5e194ec8 : SUCCESS in 1h 09m 13s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f009699fff745418c56544bfc7e944a : SUCCESS in 49m 10s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/32c78248bcea4bae863879c4f6f13959 : SUCCESS in 9m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0303c796bf4b4661abb841a8194413d1 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fa1fdcf3bf0dde0871b28c57f10c784a0e3680cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 15:52:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"68b611a4a824544d86ace10ad51b629ee65aaa37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 15:54:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6ed1d3c5242490d8232133e7c484718 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab10d814a09179c3e98bce3668550f48cef5dcab","revisions":{"ab10d814a09179c3e98bce3668550f48cef5dcab":{"kind":"REWORK","_number":1,"created":"2022-04-07 10:02:33.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/25/836925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/25/836925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/25/836925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/25/836925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/25/836925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/25/836925/1"}}},"commit":{"parents":[{"commit":"bcab43d802e651ffe046a417a21ee80b95743ff5","subject":"Prepare the final stable/ussuri release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/bcab43d802e651ffe046a417a21ee80b95743ff5"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-09-08 10:16:11.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-04-07 10:02:01.000000000","tz":120},"subject":"Add socket keepalive options for the pymemcache backend","message":"Add socket keepalive options for the pymemcache backend\n\nThis patch specifies a set of options required to setup the socket\nkeepalive feature of pymemcache (dogpile.cache) cache backend.\n\nDepends-On: https://review.opendev.org/803716\nChange-Id: I7deed1bca1ce2f715967e966d4b0cedc1f227027\n(cherry picked from commit 60495a42c9e46144dc3625cef6ede24e18245011)\n(cherry picked from commit ba251d3e2d85c1c247968c5103805ba6f144a536)\n(cherry picked from commit 02f7cc8a8fdbe8091edd561add51805cb340249f)\n(cherry picked from commit dc42ef8ed7b5804491bd1355fd8bdeb84b4ca629)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/ab10d814a09179c3e98bce3668550f48cef5dcab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/ab10d814a09179c3e98bce3668550f48cef5dcab"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
