)]}'
{"id":"openstack%2Fpuppet-oslo~853717","triplet_id":"openstack%2Fpuppet-oslo~master~Ifa79d21e5945917c5c510980b8d79847443f5282","project":"openstack/puppet-oslo","branch":"master","attention_set":{},"removed_from_attention_set":{"35067":{"account":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"last_update":"2022-08-19 01:32:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ifa79d21e5945917c5c510980b8d79847443f5282","subject":"Update .gitreview for bugfix/21.0","status":"ABANDONED","created":"2022-08-19 01:14:41.000000000","updated":"2022-08-19 01:32:55.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ec1a9df2a7c329ada827bd5f0306a95e7be20bc0","_number":853717,"virtual_id_number":853717,"owner":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-19 01:19:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-2,"date":"2022-08-19 01:29:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-19 01:19:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-19 01:29:32.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"a1107f17e8137b17d65ffea8b8b3f432b2edd6a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-08-19 01:14:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"047ab9924f29e5f13da63bfcbf6808b3c9df5782","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 01:19:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bbf4100e0cd4a609425a5583edcf808\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfca2c5e473f432f96cd78fbe9632b06 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b648a9d1586eb7e05c60126570ccaac84665f19f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-19 01:29:32.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec1a9df2a7c329ada827bd5f0306a95e7be20bc0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-08-19 01:32:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"678b5c46225b81f5e49c7a066c3f6068d5ad117e","revisions":{"678b5c46225b81f5e49c7a066c3f6068d5ad117e":{"kind":"REWORK","_number":1,"created":"2022-08-19 01:14:41.000000000","uploader":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"ref":"refs/changes/17/853717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/17/853717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/17/853717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/17/853717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/17/853717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/17/853717/1"}}},"commit":{"parents":[{"commit":"513573850b24220a51f2fe51f0bb087eefab4438","subject":"Fix typos is parameter descriptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/513573850b24220a51f2fe51f0bb087eefab4438"}]}],"author":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-08-19 01:14:23.000000000","tz":480},"committer":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-08-19 01:14:29.000000000","tz":480},"subject":"Update .gitreview for bugfix/21.0","message":"Update .gitreview for bugfix/21.0\n\nChange-Id: Ifa79d21e5945917c5c510980b8d79847443f5282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/678b5c46225b81f5e49c7a066c3f6068d5ad117e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/678b5c46225b81f5e49c7a066c3f6068d5ad117e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
