)]}'
{"id":"openstack%2Fpuppet-oslo~865458","triplet_id":"openstack%2Fpuppet-oslo~master~I5c295c747a0f19757e3b3c7eaf990b32d7e1af11","project":"openstack/puppet-oslo","branch":"master","topic":"puppet-ciphers-array","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-30 01:08:27.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I5c295c747a0f19757e3b3c7eaf990b32d7e1af11","subject":"cache: Support array value for tls_allowed_ciphers","status":"MERGED","created":"2022-11-23 18:02:50.000000000","updated":"2022-11-30 05:08:48.000000000","submitted":"2022-11-30 05:08:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865458-puppet-ciphers-array","meta_rev_id":"778e97eddae6b4042068cf0746cf29e864e0e444","_number":865458,"virtual_id_number":865458,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-30 05:08:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-11-25 13:52:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-11-29 17:03:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-11-29 17:03:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-23 20:07:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-25 13:52:02.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-11-29 17:03:04.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a9130625e742071aeb12e8cae30fb52f2d7717b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-23 18:02:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1717ee5f41d1d2513453aaf9e25ccf8994f280ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-23 18:04:25.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"68a8a414e29e2643eda0ca3623808d09c347940c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-23 18:13:55.000000000","message":"Topic set to puppet-ciphers-array","accounts_in_message":[],"_revision_number":2},{"id":"9a9abbda30945b659545a405651ceab7bc542d42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 20:07:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a459ea54dc544c8ab63d2def803a07c7\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c08dc2176e4a41969b23754528b27f13 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea3051dc5ada4d26b94a9b16ed1cfe94 : SUCCESS in 3m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c31096361484fa2b50dad45737ef7f2 : SUCCESS in 3m 23s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/470fc02a0f6f4d918630d784764254b3 : FAILURE in 9m 56s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c341145e32f74e55b2a7d2b6ffe29bfa : FAILURE in 6m 08s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd50282a44064b15a2410fd8ab5c1642 : FAILURE in 6m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbfa003eddf24c8390c3479f7f799baa : FAILURE in 5m 39s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb1491f571c949a59a2a4e0f1b7ce014 : SUCCESS in 44m 53s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c919531fea547e7b86b776644c577b0 : SUCCESS in 1h 01m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11c281a742394dfa84f21cf4031c9190 : SUCCESS in 41m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0500d685ee1457db518328d0288d390 : SUCCESS in 36m 39s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/475864d66e4241779fd5a559848b6482 : SUCCESS in 40m 46s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6bb587c7c8954488a294991c14c7ead0 : SUCCESS in 19m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5236145e2625422c90c0e0ff0f1485c7 : SUCCESS in 1h 01m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ebd23eaca2a44b2f9061e43bc1632961 : SUCCESS in 1h 14m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ffb6e219c3f4de2b33ba0f102ce8c09 : SUCCESS in 1h 08m 34s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/358d1292fc6f434581b96823a8f6edab : FAILURE in 1h 05m 31s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/74ffe33338cb481a8ef5be9527321beb : SUCCESS in 55m 53s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d07fb2ee18244fcc90a91623c9b59600 : SUCCESS in 1h 55m 59s","accounts_in_message":[],"_revision_number":2},{"id":"bcaeb407d51293027a19e39537c3ac62f8ccd20a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-24 01:41:08.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"5c449ca1d884920412d95a961ce2436632603712","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 03:39:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8ea967628d149bd87de878e66bcdd35\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49cfe39b227149658c0bdcfbb27c0b41 : SUCCESS in 3m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afdb06cd25674a33827a0ecd1ca943e3 : SUCCESS in 3m 07s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/022b2c91dd794ef8898481f39b898bd5 : SUCCESS in 3m 10s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/9119b16a3d434698b6b026ebb34b4ce4 : SUCCESS in 10m 31s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d19681bb7e2743ed8f2e732aeb6eba78 : SUCCESS in 7m 48s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f66e1ad67a44cb8b8541c05b5f86e48 : SUCCESS in 6m 49s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2f359982d5547839f594da81cc40a97 : SUCCESS in 6m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cf663a97df5406dbbf1346e81841995 : FAILURE in 17m 29s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4705d24354914fda9cdb96400c79e7e1 : FAILURE in 16m 50s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a605921a14841b59195ef0ab43eacca : FAILURE in 17m 12s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f08b55daa0e4bf39e05cf15c349fa36 : FAILURE in 16m 40s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b1b514a3c7747f1b1b109fad5e45ff7 : FAILURE in 15m 59s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/96a824a1051c4d9cae2e54d6f216ae3b : SUCCESS in 21m 59s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5024f8512034d5ca39b3fc0f8efe04a : SUCCESS in 1h 03m 29s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc0632e9da8a4c66bd08d00d12795daf : SUCCESS in 1h 20m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/39b6b540c35543f5a0f8fdc62225c861 : SUCCESS in 1h 12m 57s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d92e03e3f5048719dddda05d82ced84 : SUCCESS in 57m 20s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3c2960426f44a4ebef7481e0e7bbc8a : SUCCESS in 58m 01s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1d47411054e74a4fb9fb2d7a7b0ca119 : SUCCESS in 1h 56m 05s","accounts_in_message":[],"_revision_number":3},{"id":"b97883b34c16fb29988da93b77e93e28f226df74","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-25 13:52:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"dc0c7d32a65f561c07c8bf29a07be488f79788ea","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-29 17:03:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5ae06f45a3bc4f732efb0d2220d18c5a6bf58c00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 17:04:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a830523bc5ccf03fcad990d008f92e5b9c7f03a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 18:48:47.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8935d72780e945258ec20493a75afddd\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0935047254f44319b69dac087ebf7719 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f41251dee4164bb9994b9d948b75dba2 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01c3028ec5f944a9a80395138fce8872 : SUCCESS in 3m 14s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a39b5b4c4194335b8b838fd7664caec : SUCCESS in 12m 00s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b20868aae5384c8e91ae2b6059d75a06 : SUCCESS in 8m 39s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b852ad91f1d4299af8096c233e12b1c : SUCCESS in 8m 15s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/85eeb15e9e024b009a073114b72a53b9 : SUCCESS in 1h 04m 25s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eeb397b3b0574f1796c5a9ce847b8adb : SUCCESS in 1h 22m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0786e35366f7440aaa3714bd9ae8e6f7 : SUCCESS in 1h 02m 15s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffbb438a4dfd4e04b16572b6dddf891b : FAILURE in 54m 50s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/931f1f76df7a42d88abcf2bb3e3c7361 : SUCCESS in 57m 23s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b93da43cada34fd488807c1cc7aa13c0 : SUCCESS in 1h 41m 10s","accounts_in_message":[],"_revision_number":3},{"id":"9122f3eadb93b3e0d09ccb3ef703eb8c60db1cf3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-30 01:08:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9192a46371439fe3b32361f6abc371baab893c27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 03:06:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/298cf74db73149d08c68a69f7ff71b39\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4a9f744b191458f9a7123cca5758bfa : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6ff13bf6f5143648d06369b01cc5b4a : SUCCESS in 3m 25s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95069b6050a54448bcfe57ddc274b0ba : SUCCESS in 2m 36s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d22f8fe4b5e486aad4b41a1c92b187a : SUCCESS in 10m 52s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03a5287ca1bf4e95b3b292ebb3dca7e2 : SUCCESS in 5m 44s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a29ae457b05940bfbba9e64e9ca9aa06 : SUCCESS in 5m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97afd22092754019a205afcfa0dcbe36 : SUCCESS in 5m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06a03fde7686450697be8695bc67f951 : FAILURE in 18m 29s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b24434f88b2441ba2e4ec84e02a14ad : FAILURE in 21m 07s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01fd17ccce9e463fa71291f7d5ce9174 : FAILURE in 17m 06s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ed596865b59445b8c71aa84a249bd65 : FAILURE in 17m 17s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79cf4a81bb7d4148b3935ecb5f087661 : FAILURE in 16m 31s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/50bbe99a7484419689cae3491a23b1be : SUCCESS in 19m 20s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9be27b96d05d4d13863a5fa7b03f63a6 : SUCCESS in 58m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a96435fd42a4908b78df8e20b540f9f : SUCCESS in 1h 12m 28s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/19183962dab046df89a8b9869618d4ac : SUCCESS in 1h 03m 59s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/07e8b9fb0d2d47d88ac6d12dd990be6e : SUCCESS in 57m 25s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb350e1c846f4575afd05910e87b1f81 : SUCCESS in 53m 25s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b16d48b2684843a3aaf0ce936f9863fe : SUCCESS in 1h 51m 49s","accounts_in_message":[],"_revision_number":3},{"id":"230148a8b3c769512f7756fadf94e426c9a068a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 03:07:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"37b4d07134e8f30438ea4c81b6c4c0547dc1f755","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 05:08:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/775bbcd99e204fd2bd74e749d2fdfe9e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05169441f2384c36905aedd655c0e4d7 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/679198ae76604622805df1b375bdb646 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/877cc701bfe14feb93310c9dbb051a6f : SUCCESS in 2m 52s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/73a5ec3c4c4a4f2c960d2a67251589c8 : SUCCESS in 9m 46s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb552f48cd2943d39101ede2bf10452e : SUCCESS in 6m 22s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6af8859f42b46a3863bb295e5a0fb1e : SUCCESS in 6m 06s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb2517fdacdc40798aeff6e3781413b2 : SUCCESS in 1h 10m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4a8d13652d24f11b9c3fd84f18c89e6 : SUCCESS in 1h 15m 22s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/030ba309a1704b51b4118afdf903e1bf : SUCCESS in 1h 07m 09s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e75749ded89548e092117e9b8622d169 : SUCCESS in 57m 09s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ebdacefb2a974b36b65e71471acf243c : SUCCESS in 53m 30s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/62e44196c5b8468db8ee07621a29cfed : SUCCESS in 1h 55m 11s","accounts_in_message":[],"_revision_number":3},{"id":"778e97eddae6b4042068cf0746cf29e864e0e444","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 05:08:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"053264927508064e7c832c88b080dec8b85a290e","revisions":{"58afe237f92cf81bcae7be08cd6baa3a703622e9":{"kind":"REWORK","_number":1,"created":"2022-11-23 18:02:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/865458/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/58/865458/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/58/865458/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/58/865458/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/58/865458/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/58/865458/1"}}},"commit":{"parents":[{"commit":"7040315eca17db5a068c7921517286184bf961b7","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/7040315eca17db5a068c7921517286184bf961b7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 18:00:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 18:00:52.000000000","tz":540},"subject":"cache: Support array value for tls_allowed_ciphers","message":"cache: Support array value for tls_allowed_ciphers\n\nThe [oslo_cache] tls_allowed_ciphers option takes a colon-seprated\nlist. This change allows usage of a native array value to set\nthe value so that users can more easily compose up the value without\nconsidering the actual value format.\n\nChange-Id: I5c295c747a0f19757e3b3c7eaf990b32d7e1af11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/58afe237f92cf81bcae7be08cd6baa3a703622e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/58afe237f92cf81bcae7be08cd6baa3a703622e9"}]},"branch":"refs/heads/master"},"38bf0e78d085afba9b35c3c45ad8995cf47e0582":{"kind":"NO_CHANGE","_number":2,"created":"2022-11-23 18:04:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/865458/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/58/865458/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/58/865458/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/58/865458/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/58/865458/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/58/865458/2"}}},"commit":{"parents":[{"commit":"7040315eca17db5a068c7921517286184bf961b7","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/7040315eca17db5a068c7921517286184bf961b7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 18:00:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 18:04:14.000000000","tz":540},"subject":"cache: Support array value for tls_allowed_ciphers","message":"cache: Support array value for tls_allowed_ciphers\n\nThe [oslo_cache] tls_allowed_ciphers option takes a colon-seprated\nlist. This change allows usage of a native array value to set\nthe value so that users can more easily compose up the value without\nconsidering the actual value format.\n\nChange-Id: I5c295c747a0f19757e3b3c7eaf990b32d7e1af11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/38bf0e78d085afba9b35c3c45ad8995cf47e0582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/38bf0e78d085afba9b35c3c45ad8995cf47e0582"}]},"branch":"refs/heads/master"},"053264927508064e7c832c88b080dec8b85a290e":{"kind":"REWORK","_number":3,"created":"2022-11-24 01:41:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/865458/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/58/865458/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/58/865458/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/58/865458/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/58/865458/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/58/865458/3"}}},"commit":{"parents":[{"commit":"7040315eca17db5a068c7921517286184bf961b7","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/7040315eca17db5a068c7921517286184bf961b7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 18:00:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-24 01:41:08.000000000","tz":0},"subject":"cache: Support array value for tls_allowed_ciphers","message":"cache: Support array value for tls_allowed_ciphers\n\nThe [oslo_cache] tls_allowed_ciphers option takes a colon-seprated\nlist. This change allows usage of a native array value to set\nthe value so that users can more easily compose up the value without\nconsidering the actual value format.\n\nChange-Id: I5c295c747a0f19757e3b3c7eaf990b32d7e1af11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/053264927508064e7c832c88b080dec8b85a290e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/053264927508064e7c832c88b080dec8b85a290e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
