)]}'
{"id":"openstack%2Fpuppet-oslo~883631","triplet_id":"openstack%2Fpuppet-oslo~stable%2Fzed~I778fdc2dffcc50cf447634dbe56eee289c2b153c","project":"openstack/puppet-oslo","branch":"stable/zed","topic":"remove-tripleo-zed","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-05-22 05:27:47.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I778fdc2dffcc50cf447634dbe56eee289c2b153c","subject":"Zed-only: Remove TripleO job","status":"MERGED","created":"2023-05-19 04:50:36.000000000","updated":"2023-05-22 08:10:42.000000000","submitted":"2023-05-22 08:10:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883631-remove-tripleo-zed","meta_rev_id":"18d52a93fcb93409d4c96ecb62021d2dce324ba5","_number":883631,"virtual_id_number":883631,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-22 08:10:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":1,"date":"2023-05-19 10:15:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2023-05-22 03:37:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2023-05-22 03:37:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-19 06:19:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-19 10:15:39.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"}],"messages":[{"id":"7d89dc9fae07f4937c6f05a8846bfe56ce14ebe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-05-19 04:50:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20d06ac3b15e83d8b3833c53f923212399853f90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 06:19:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d97a2f35055642c2aaa21e9982a5484a\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a5bfb17ac684986a8052e5598c75a3e : SUCCESS in 2m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fc85b7b5e8c4aa6b84255992006af79 : SUCCESS in 3m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ff7ea0021c2454a899d2987a22dec77 : SUCCESS in 3m 40s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/23d8a9bc7145434ca9b907bf0163ca81 : SUCCESS in 7m 39s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bc24253ad784b3abe209a93e9b3633f : SUCCESS in 5m 31s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cce8c90365ab43a3b3d08a9ab66e3c05 : SUCCESS in 7m 08s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6da07080f2f64025a2914164e44c10b2 : SUCCESS in 5m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72b67d86a7354a33a77afee2e3ef9259 : FAILURE in 16m 37s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee14c5fdede1485aa28c23c7e19d5e92 : FAILURE in 16m 43s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d10f3e99f7f4b6891911925c9fc89bc : FAILURE in 17m 56s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ceb61da21ea1462da41e38b2dbd1a140 : FAILURE in 17m 41s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a610d92e553f4e1aba8e9bb44d188c54 : FAILURE in 16m 22s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/28b9560fcae6431bb3c4faf80297e891 : SUCCESS in 20m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea8e88d9d8b143a3923ec150701aad96 : SUCCESS in 1h 00m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/110ca01d0c06410fb651785755b4133f : SUCCESS in 1h 21m 22s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/47ed12ee1d424d859bdb703f251e1be2 : SUCCESS in 49m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5dcf3337f2164e3bb842e08a14f25948 : SUCCESS in 56m 35s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1b3d80d6de94cc58d8285bab23f3a4e : SUCCESS in 46m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a0270c72e0b750f4f4e74fdae5686031f084d1ca","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2023-05-19 10:15:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f3ea709f939e3b3ac80c43b1bf7de61a232e142c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-05-22 03:37:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c634f559e44150af55e267c637dcbfef2c0eb53d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 03:39:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6acc15d4d724d6c48c8fcb5f265214ea337e2de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 04:57:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b972342cea37492788435eda2e8f73d3\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6162fcfbbc0a4a45a51afd7557d8b71b : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24051f88b35548d1a6001c7da1e539f0 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de9baac509da43cc8268606ab44d3749 : SUCCESS in 3m 18s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/75b7a3a37b6f4e8588604c8dee43905d : SUCCESS in 9m 19s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ae86f3d4db2483ebedcc87888c37bf8 : SUCCESS in 5m 48s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c4ac38cd09343cdac3f1a8123087004 : SUCCESS in 6m 41s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5257229bc9ed4812933e451af99d90c2 : FAILURE in 55m 02s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/680c55b68e8641da8159337638c1f4e6 : SUCCESS in 1h 13m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6056bd220e054ca790245514f86f5f8b : SUCCESS in 58m 26s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb3c1303e229421bb266da223d6dbaa7 : SUCCESS in 50m 57s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9363031de0b4173bd4932582dd97e1b : SUCCESS in 40m 33s","accounts_in_message":[],"_revision_number":1},{"id":"d83fc3ab2dfd4256f4db6e4aec0844e9240849e8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-05-22 05:27:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a5e34b95e4d7d0bfe0d735d4d804914871712013","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 06:45:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fae0730327b34def8ba70a057b1eb518\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/411dd85ff28d400c934f4bdaf305ef8d : SUCCESS in 2m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d6257976aa14154ab4e5444a85a7429 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1489695f2ab1497693f15453e60652e9 : SUCCESS in 2m 50s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ceba11b56e248ffb6ab64ab021e67e5 : SUCCESS in 8m 47s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2fab97574634125a03cce045bd21db4 : SUCCESS in 5m 46s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9505008abe9d495d853f34068f8aec2d : SUCCESS in 7m 34s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/570b24bd62b44e7291b997d362fd96c8 : SUCCESS in 5m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c85caab8e9b4b3490a2d9c350854ae1 : FAILURE in 17m 34s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81a16b199c80446988fe197b91b299de : FAILURE in 16m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd55a181c4084e239cee9adbd93c9a00 : FAILURE in 17m 33s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef2558c10af5433998e4ef1b9a534d67 : FAILURE in 16m 35s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06a0a162a7a04e79a8902cfbcf292005 : FAILURE in 15m 25s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/51cc14392c404e809f562b4e6eb5c6d2 : SUCCESS in 15m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf3ac4001f9e4fe7858fdab1d50d699f : SUCCESS in 49m 34s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fee353a1ae047ee9c9cd3167bc141b4 : SUCCESS in 58m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a3d2d4713694e768191d3dcd152d893 : SUCCESS in 1h 00m 16s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/812c47042018406ebe1e4fa054747c30 : SUCCESS in 53m 03s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bee6f45364fd4e84ab9d429c8e9dae9b : SUCCESS in 44m 10s","accounts_in_message":[],"_revision_number":1},{"id":"7cd788cf6005e945a54355e2c97e2244af9723d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 06:46:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b3ea1e81479cd5b8f01817c29159ba122ae6e0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 08:10:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f24e491229604727a3777c0a17106d1e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4be4918940e14d3bab5cb4bdc8202a88 : SUCCESS in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c4db3625ab84aa6a42afe794dfbc36d : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db815bbd2fbb4952a134484acd168949 : SUCCESS in 2m 53s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9e8d32945d8492d8f413b19abaf563d : SUCCESS in 9m 10s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec62216d49af4ea79c5dd5ddbba37f66 : SUCCESS in 6m 36s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c1159437b4f45c59f83c89ad3237611 : SUCCESS in 7m 47s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e890d1ba2772419abd98ba8992044fa8 : SUCCESS in 52m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1114665e1f6419983438c604e26be31 : SUCCESS in 1h 23m 18s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/33cbc952c0f94eb291e1a640637be5c0 : SUCCESS in 55m 03s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a936f98e3b484cabac876229d65f65f3 : SUCCESS in 58m 19s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/04d03453c0874103a3cc4bd816f30651 : SUCCESS in 48m 24s","accounts_in_message":[],"_revision_number":1},{"id":"18d52a93fcb93409d4c96ecb62021d2dce324ba5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 08:10:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f712954d43412b4df839a29a56e29c75c73ceac6","revisions":{"f712954d43412b4df839a29a56e29c75c73ceac6":{"kind":"REWORK","_number":1,"created":"2023-05-19 04:50:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/883631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/31/883631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/31/883631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/31/883631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/31/883631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/31/883631/1"}}},"commit":{"parents":[{"commit":"3007fcf8301894bac198fa584b659caaaea9237b","subject":"Fox wrong format of [cache] backend_argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3007fcf8301894bac198fa584b659caaaea9237b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-05-19 04:50:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-05-19 04:50:27.000000000","tz":540},"subject":"Zed-only: Remove TripleO job","message":"Zed-only: Remove TripleO job\n\n... because the stable/zed branch of TripleO repos is no longer\nmaintained.\n\nChange-Id: I778fdc2dffcc50cf447634dbe56eee289c2b153c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f712954d43412b4df839a29a56e29c75c73ceac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/f712954d43412b4df839a29a56e29c75c73ceac6"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
