)]}'
{"id":"openstack%2Fpuppet-oslo~885938","triplet_id":"openstack%2Fpuppet-oslo~master~I58de785f83e11d3db2393bdc8dcd240f37e15fdd","project":"openstack/puppet-oslo","branch":"master","topic":"replace-validate_legacy","attention_set":{},"removed_from_attention_set":{"7888":{"account":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"last_update":"2023-06-19 07:52:41.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_7888\u003e by clicking the attention icon","reason_account":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-06-19 09:31:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I58de785f83e11d3db2393bdc8dcd240f37e15fdd","subject":"remove validate_legacy (Booleans only)","status":"MERGED","created":"2023-06-13 07:13:41.000000000","updated":"2023-06-19 09:31:47.000000000","submitted":"2023-06-19 09:31:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885938-replace-validate_legacy","meta_rev_id":"0cb62ba3c54aec1498ce40faeeaec07fd7ecd882","_number":885938,"virtual_id_number":885938,"owner":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-19 09:31:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-06-19 07:50:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-06-16 21:21:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-06-19 07:50:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-13 08:21:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-15 18:24:48.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"},{"updated":"2023-06-16 21:21:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-06-19 07:50:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"f4fae97ed69ac9f5196636d892d4375cc6b7b162","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-13 07:13:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18edb22a6aaf96c4262707e2517685c1d2831fba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 08:21:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7da173b3ef7b462183879d93657d13d2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca478bb27dfe4339a815c0527dbe973c : SUCCESS in 2m 38s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07b30d216a8c44a0913af8591f0e27f1 : SUCCESS in 2m 20s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/513b56579b3b4124b94748c973346ed4 : SUCCESS in 4m 50s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9654592faea42ab84d56b37581d8fc7 : FAILURE in 8m 52s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebb1fe837bcc4ee28baa058e491b4432 : FAILURE in 7m 52s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0c01062aaf64772929c805ac5821af9 : FAILURE in 12m 08s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85b632f5044340958af81a5c40bf6748 : FAILURE in 8m 15s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a8dcceb1eaa43708e2f19e1a8f90b88 : FAILURE in 8m 54s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca8c892376084481a2fa05e6c2473714 : FAILURE in 5m 46s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bc7a2c3acd34d3db1984bf0abe4569c : FAILURE in 7m 31s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a06bbf9f0a34ceebd36b9d9632b2a5c : FAILURE in 5m 31s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb19fc71dd704a9ea7a3d7f7478bd3a8 : FAILURE in 7m 34s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/666cb68d46da48b1aae715e917c86483 : SUCCESS in 21m 48s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6bc42a7b81543b79129681865b8c469 : SUCCESS in 59m 07s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ba8ea44fe5c473e89f8006fd9b7dbab : SUCCESS in 1h 05m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a527c6cff494b47abeaf94caf6484b8 : SUCCESS in 1h 04m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ddde6c25d8c4e679946f09968463379 : SUCCESS in 49m 59s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b46d5801ece4f0b8d49fb8acceae74f : SUCCESS in 51m 41s","accounts_in_message":[],"_revision_number":1},{"id":"19bb6c98a7795e4685161c5673394a9fe16e4c7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-13 08:35:45.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"45070030b06ea0662729e7f791214fc8bc212e75","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-13 08:36:22.000000000","message":"Topic dev_validate_legacy removed","accounts_in_message":[],"_revision_number":2},{"id":"eb3582ce3e3f5ed5de347709e242a96c45e4c302","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-13 08:36:30.000000000","message":"Topic set to replace-validate_legacy","accounts_in_message":[],"_revision_number":2},{"id":"2ae42de456de348a3144afd912220d3ef32e8842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 10:03:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/10ce9e8e98c944a5b6aff491dc60a092\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ceebcd1890104d07b4166b1807841e34 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e7409c7b46a4875993a509797026e28 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/1675e622c4f143c6a71d0e3c55d36dfa : SUCCESS in 4m 28s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6699313cd01a4727818f1e446191355d : FAILURE in 11m 45s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3155c16844e423f8c545b537d350316 : FAILURE in 5m 16s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/98b75716edec4797a2b6b64762c9ffce : FAILURE in 8m 10s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/118464dd171541aab767b16ac780a4b3 : FAILURE in 5m 22s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05e62e25fb7449ff8b611731156cf256 : FAILURE in 6m 30s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffca5e1660e448ecb6b4ec7216cdb4cc : FAILURE in 7m 37s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/942116f411d849a8baf07d3e3a4c8f9c : FAILURE in 6m 44s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db055178df964e9eb849704dbc5f1c90 : FAILURE in 9m 53s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c969783d58c49c4b73d80c69df59bac : FAILURE in 5m 39s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f36c8f8190104cdabe397167ed8242d5 : SUCCESS in 23m 03s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef520b1fd89048be9a39d9a0db31161e : SUCCESS in 1h 09m 19s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc8d381f513e4a328d81ce001d0b295d : SUCCESS in 1h 12m 15s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9abf3c87f6124406a333f6b725e6af61 : SUCCESS in 1h 17m 52s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b144a5b8b76452b8bac024fb095b343 : SUCCESS in 57m 29s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4191d3d04344b0099f65642a57497d6 : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":2},{"id":"6a4f673f6bf1c6b1473dbfede661204c7752f4bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-13 11:00:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4bcb12a78bf994fbc5c47faf83a41447fd12cc7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-13 12:15:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"61fa105e8a8124b49c0642947823c23ef81ff7cc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-13 12:40:45.000000000","message":"Topic dev_validate_legacy removed","accounts_in_message":[],"_revision_number":4},{"id":"1ac57f2943bd3d186f1abbb689630c968e445665","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-13 12:40:53.000000000","message":"Topic set to replace-validate_legacy","accounts_in_message":[],"_revision_number":4},{"id":"224ff0590cfa4b22ae52c1b624b0ad7fa9101d12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 13:34:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbd2a8009fbc4a0f94d7b11bd5fd8103\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/344aee69697442ea8945d4c8ba5b0025 : SUCCESS in 2m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/628ec9df095b472e85f1c731c18edcf9 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/efb06cffcd1f43c5bfc703a14c92043b : SUCCESS in 5m 14s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/787b1ed2b42447a6b382e1e7fc8bd17e : SUCCESS in 11m 03s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b398618067e043ae84d670a873ac1e29 : SUCCESS in 7m 17s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/4459998e0a8640a6adae001021926dd1 : SUCCESS in 9m 49s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ada89ec61e514685a0ea41dc601209a9 : SUCCESS in 6m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75dee5fdc6d34b0fa2923318b0c53ff6 : FAILURE in 6m 37s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9ed1632ec544a54bb3e113335d0e033 : FAILURE in 6m 51s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0c3a03d1c88475983500552664165c5 : FAILURE in 6m 11s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18b4a776b5a2437eabeec20f440caabe : FAILURE in 7m 29s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddf87385dd66430db8d7f35d67e15f0f : FAILURE in 6m 51s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad9bf51af9a44f818a75de6a9a35888c : SUCCESS in 20m 27s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/027a560f66bc4512b422214d41803c2a : SUCCESS in 59m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1db4fb7f3c1744a98d973265045b23fd : SUCCESS in 1h 13m 35s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/56134c96be7f481a9c9b15626c9d5e3c : SUCCESS in 1h 05m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/950d77494d044b8aae28c38c606f8979 : SUCCESS in 1h 01m 38s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a90ee10d07c49f382f6edf777ca7776 : FAILURE in 56m 27s","accounts_in_message":[],"_revision_number":4},{"id":"af6e77c4b096321cd9b8cf91ed938bce25b5b6cc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-06-15 18:24:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1ead32dd369102193d371a7bafab9a47c51df8ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-15 20:18:05.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5975f378588abd6555b168c5d2ae5a58dd835ddd","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-15 20:18:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"73747ad74a2b2d8a56ac1a248dadfc76bf09a34e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-15 20:19:49.000000000","message":"Topic dev_validate_legacy removed","accounts_in_message":[],"_revision_number":5},{"id":"16e47cab6bafabc7164442d6894f7cda5cc2ea95","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2023-06-15 20:19:53.000000000","message":"Topic set to replace-validate_legacy","accounts_in_message":[],"_revision_number":5},{"id":"a206dcda0af717e0e8b2546c9c7c198119054978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 21:34:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2adc3e250365448280dfc0f754d97eb4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3316a12acc18470191b53653d4b52223 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2907bec26e074c208cf45a2f14576293 : SUCCESS in 2m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/96f07cf5f5c4476ead91c56916357ad1 : SUCCESS in 5m 32s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/170c9d6f35cc4a2eb4b0f91b0f7eeb14 : SUCCESS in 9m 18s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3de2ac5b479b413a8d0aa41271d65ca8 : SUCCESS in 5m 47s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3a5caeaef2d44f985d1cded2ac2417f : SUCCESS in 9m 06s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d817d8206e54aefb11a0cee3c58d494 : SUCCESS in 6m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdb24508d90a48c887c2b22c0f1c297b : FAILURE in 6m 14s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e4a70189b3c429b9ad24b657c99697b : FAILURE in 6m 14s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7206bb2f5ae2410ca09421a79aaafce9 : FAILURE in 6m 36s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9e1c719c985450094fdc2fad2c2ab7f : FAILURE in 6m 15s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/437f6016daab42b7a52fc8c09766280d : FAILURE in 6m 31s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a73a1586e22d487e98626427fcff6533 : SUCCESS in 20m 35s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/67dc8be076e846caa67835bf3ea943cf : SUCCESS in 59m 18s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/16d72af54de44d02bae1e40400cf59ec : SUCCESS in 1h 11m 07s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbde9599806b4edbae15a0131a0855ce : SUCCESS in 1h 07m 10s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d990f0d894146e2b803b0dc8bb2e5f4 : SUCCESS in 56m 22s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4606e15c5d544332bd57bc4f27f2f6e0 : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":5},{"id":"57dadd369db35dd8aa7b4b1840693336e78ac5b8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-06-16 21:21:03.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1cb9e8834b47ea4aec23b047170c80277cfe820f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-19 07:50:50.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2ef11aa4cf982da722f305d2b3ccb77a8be9c582","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 07:51:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"dc02cb45512091574b9ecaa3d127dfa442aac8b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 09:31:46.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe22d2fcf2e24e84a0f0a6dd238438ff\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8d3a6ff6bfb48dc8bb91c885669357f : SUCCESS in 2m 25s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ff32f746a1a49978a1fdc78d1a030ba : SUCCESS in 2m 38s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/13c7acf52a474c1c9d4b0c2b97c6ddfe : SUCCESS in 9m 34s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f61cbc89776a43b3a86a4049c9963e5a : SUCCESS in 6m 12s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5f9eef7069a4e46b6e6265767f33e5c : SUCCESS in 9m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d3067b52eec47a68ca01a58e106f572 : SUCCESS in 1h 02m 34s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/881aba629aee4f0e9c028cf3b933c10d : SUCCESS in 1h 33m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2db04f24817b4dd7a2668a61e9f33077 : SUCCESS in 1h 08m 51s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8fac139c8d3466b8f621db66272664e : SUCCESS in 59m 07s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/81159dab288a4461b3091d4d2043367c : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":5},{"id":"0cb62ba3c54aec1498ce40faeeaec07fd7ecd882","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 09:31:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fd712632841326e0e3b0ae38c80144783c8bcde7","revisions":{"15d275dc2d80ab55afa3df7bb31e45dc67c200f4":{"kind":"REWORK","_number":1,"created":"2023-06-13 07:13:41.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/38/885938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/38/885938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/1"}}},"commit":{"parents":[{"commit":"471514fa533e5c42a94919fe884fa4a7b95751bc","subject":"add spec test for type Oslo::Dbconn","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/471514fa533e5c42a94919fe884fa4a7b95751bc"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-13 06:55:46.000000000","tz":120},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-13 06:55:46.000000000","tz":120},"subject":"remove validate_legacy throughout the module","message":"remove validate_legacy throughout the module\n\nChange-Id: I58de785f83e11d3db2393bdc8dcd240f37e15fdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/15d275dc2d80ab55afa3df7bb31e45dc67c200f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/15d275dc2d80ab55afa3df7bb31e45dc67c200f4"}]},"branch":"refs/heads/master"},"068c5195c60efd54617ae3ea05f8e7b76f3ec97b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-13 08:35:45.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/38/885938/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/38/885938/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/2"}}},"commit":{"parents":[{"commit":"72bd48c9506c563f58ba2e3a94be1ea63cb67a16","subject":"Remove support for Puppet 6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/72bd48c9506c563f58ba2e3a94be1ea63cb67a16"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-13 06:55:46.000000000","tz":120},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-13 08:34:25.000000000","tz":120},"subject":"remove validate_legacy throughout the module","message":"remove validate_legacy throughout the module\n\nChange-Id: I58de785f83e11d3db2393bdc8dcd240f37e15fdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/068c5195c60efd54617ae3ea05f8e7b76f3ec97b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/068c5195c60efd54617ae3ea05f8e7b76f3ec97b"}]},"branch":"refs/heads/master"},"96caf7127d84d8e8e14bdf572bfed2d54f3bade1":{"kind":"REWORK","_number":3,"created":"2023-06-13 11:00:00.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/38/885938/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/38/885938/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/3"}}},"commit":{"parents":[{"commit":"72bd48c9506c563f58ba2e3a94be1ea63cb67a16","subject":"Remove support for Puppet 6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/72bd48c9506c563f58ba2e3a94be1ea63cb67a16"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-13 06:55:46.000000000","tz":120},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-13 10:59:23.000000000","tz":120},"subject":"remove validate_legacy throughout the module","message":"remove validate_legacy throughout the module\n\nChange-Id: I58de785f83e11d3db2393bdc8dcd240f37e15fdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/96caf7127d84d8e8e14bdf572bfed2d54f3bade1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/96caf7127d84d8e8e14bdf572bfed2d54f3bade1"}]},"branch":"refs/heads/master"},"b68072449cdecf566e36bbe48a5f71f306bafe9d":{"kind":"REWORK","_number":4,"created":"2023-06-13 12:15:27.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/38/885938/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/38/885938/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/4"}}},"commit":{"parents":[{"commit":"72bd48c9506c563f58ba2e3a94be1ea63cb67a16","subject":"Remove support for Puppet 6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/72bd48c9506c563f58ba2e3a94be1ea63cb67a16"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-13 12:14:08.000000000","tz":120},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-13 12:14:08.000000000","tz":120},"subject":"remove validate_legacy (Booleans only)","message":"remove validate_legacy (Booleans only)\n\nChange-Id: I58de785f83e11d3db2393bdc8dcd240f37e15fdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b68072449cdecf566e36bbe48a5f71f306bafe9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b68072449cdecf566e36bbe48a5f71f306bafe9d"}]},"branch":"refs/heads/master"},"fd712632841326e0e3b0ae38c80144783c8bcde7":{"kind":"REWORK","_number":5,"created":"2023-06-15 20:18:05.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/38/885938/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/38/885938/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/38/885938/5"}}},"commit":{"parents":[{"commit":"72bd48c9506c563f58ba2e3a94be1ea63cb67a16","subject":"Remove support for Puppet 6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/72bd48c9506c563f58ba2e3a94be1ea63cb67a16"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-13 12:14:08.000000000","tz":120},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2023-06-15 20:17:35.000000000","tz":120},"subject":"remove validate_legacy (Booleans only)","message":"remove validate_legacy (Booleans only)\n\nChange-Id: I58de785f83e11d3db2393bdc8dcd240f37e15fdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/fd712632841326e0e3b0ae38c80144783c8bcde7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/fd712632841326e0e3b0ae38c80144783c8bcde7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
