)]}'
{"id":"openstack%2Fpuppet-oslo~889911","triplet_id":"openstack%2Fpuppet-oslo~master~Id78c6c1553973b57b4b9e7edaf8d029d833f4722","project":"openstack/puppet-oslo","branch":"master","topic":"db-ensure_packages","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-08-02 11:01:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Id78c6c1553973b57b4b9e7edaf8d029d833f4722","subject":"db: Use ensure_packages to manage backend package","status":"MERGED","created":"2023-07-28 04:14:37.000000000","updated":"2023-08-02 13:58:29.000000000","submitted":"2023-08-02 13:58:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889911-db-ensure_packages","meta_rev_id":"9a9c1b5bd6519317520c125727a48485276648f1","_number":889911,"virtual_id_number":889911,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-02 13:58:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-07-28 14:43:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-08-02 08:27:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-08-02 08:27:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-28 06:08:47.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"},{"updated":"2023-07-28 06:11:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-28 14:43:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-08-02 08:27:52.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"c255919f2e3c7ce5d58d4d0d7bb31ef191c8bc89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-28 04:14:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6059ab19120b195b2d1b9e41b7d44f699270bae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-28 04:15:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7bba3e456ffca90ac02c89612529e4035e5eb828","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-07-28 06:08:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dac8bc4881a18004c50ea5050374ec167f7c252d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 06:11:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9179c55921234d7489505899c5cb632e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ba00125c0df4d8ea3d6592a1bc7c547 : SUCCESS in 2m 55s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5295d2c4c179468aa8530aa43e54a6a5 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/27fc0ef091af452f991192509d9dfc1b : SUCCESS in 5m 54s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6aa08463ff3f471eb90d6831031698fc : SUCCESS in 9m 14s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b679524b7d946d09a44511e9b94409e : SUCCESS in 5m 46s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/677b249959b34fc584ed07b251c3f8ac : FAILURE in 10m 10s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/312498cafbd94ce9ace6d355b7022459 : SUCCESS in 7m 22s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41c3b66304e44bc68398e3b7dfca5519 : SUCCESS in 52m 05s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/950c8084ca5046f692a86e42848d3277 : SUCCESS in 1h 06m 55s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92dbdc7f6ae54fe49804320380270bba : SUCCESS in 54m 23s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e75bb8d616c48ac87fba9be64ce79ac : SUCCESS in 44m 15s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20eb1cd82be94c76a87c1eb15a791fce : SUCCESS in 37m 17s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ada875e48d641a09d49975fcac3c445 : SUCCESS in 56m 39s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f15545672c746ba884eee42173512aa : SUCCESS in 1h 47m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e342f9a788f45baa65e6eb17efa705d : SUCCESS in 1h 10m 43s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a97b2b2aadbb43a4bacfbe8c2faa2b2a : SUCCESS in 55m 33s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff82c9c203814f398123267bc5885c10 : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":2},{"id":"13b4231b0451992f7a879e33ab2e6d67610b29dd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-07-28 14:43:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"414fa606515776895cad593432ae719a47130a1c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-08-02 08:27:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d1ea0d84ca4c31bf758a34b4ed422df3d662919f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 08:28:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ad1cc05c41e7e177101a634568d1f3785ba2b38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 09:48:01.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0912ec1ed04949e89f19f9ed5e648b3b\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bbf9cff52a749f7987fe2c23e99b9c5 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89879001693942fa87a71e83c21fb905 : SUCCESS in 3m 12s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f7366e29f5d400f8272cc10f57f1e97 : SUCCESS in 11m 08s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/443f2cbf65754e40956cb399dbd5a19e : SUCCESS in 7m 58s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ee50b2c616f4b06ab7debf3c93b30a2 : SUCCESS in 9m 04s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0a360510275408282d5d2ee571e98ae : SUCCESS in 56m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab34fdd41f88436a9c395171f824d6b2 : SUCCESS in 57m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e80539c64094ead81389b38354683a4 : SUCCESS in 1h 07m 18s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/58eb5c12ba6b4419bcae9f77ef9c59b1 : FAILURE in 57m 47s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bad53b0daa9c43df860e2bbe8e674486 : SUCCESS in 55m 44s","accounts_in_message":[],"_revision_number":2},{"id":"2eca838d72362e613dcaa509e8202c9c87e87485","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-02 11:01:17.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2b576b380a48e32f3bd36b035622c825a4b696d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 12:25:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efbb63b89fda48c9a50f6bb51e96ce5d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02622409675f4b718cca064dbd7b05e7 : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee9f442be4f641dc985a7fe27d37ae65 : SUCCESS in 2m 34s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/53e8738fb1c04e8aa289520511b1903c : SUCCESS in 5m 10s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d356cbcc0b841ec852c1016da589177 : SUCCESS in 9m 27s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbab5c1044804c4282866ac3d5b2d998 : SUCCESS in 5m 50s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/aab5392151a1493ea160085ca3330f06 : SUCCESS in 9m 08s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5cd871c52914480e84ed20ec5bc86ee3 : SUCCESS in 6m 58s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f791f7d3bbb347a2997b8946d7cd2678 : SUCCESS in 46m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a50c21f8ada4119aabcc535f3a1a348 : SUCCESS in 52m 25s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9472ef780cef41ddafb22cb6d4829659 : SUCCESS in 56m 23s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a0bbf30d56f4a29ab8d5d7cccb9c60c : SUCCESS in 40m 22s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a803239a9a64ba49e00d312713ba6ea : SUCCESS in 38m 29s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/be768171e0d747f29bd73b1b94312bbc : SUCCESS in 1h 09m 02s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/40b75c7b221a42bfaf78d8dd9a0989e1 : SUCCESS in 1h 18m 46s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/67d457529fb141eda607eb1fce4e5a2b : SUCCESS in 1h 09m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cdb98e2265594fe1ac88096146ef52f0 : SUCCESS in 1h 00m 23s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3698329a1e4b4abb8d03d65f0bc1a12e : SUCCESS in 53m 39s","accounts_in_message":[],"_revision_number":2},{"id":"5788509c12bba362c51c9bedc8d8d7144138d83e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 12:26:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f7a50425dd3c25522d21913e2d52144291dedd83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 13:58:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ef962da1785474dbd70d6e6f234ba8c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b414699a87945068e6fb21e0447f813 : SUCCESS in 2m 24s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f19b4ddf0ec44ba6b37e1df9224ca5eb : SUCCESS in 2m 47s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/10c0ac7f30824fc49a4c0cde8d80c1be : SUCCESS in 8m 51s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afc8614f881448a1ac83bc5537553ae9 : SUCCESS in 7m 00s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/942c9da13dc642b084aa98b0c4f2bf8d : SUCCESS in 9m 22s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/692de9f8390748c791ad33b77d02ff91 : SUCCESS in 1h 01m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/981e2be0df784c0e82b2be595e777949 : SUCCESS in 1h 27m 32s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae901add89244cfd8599fa0baa58ddb9 : SUCCESS in 1h 14m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e3cfb09a25b46bea354c9b60e8d2eea : SUCCESS in 57m 40s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1ab9a72f6724905a839a8065bd5bbbf : SUCCESS in 1h 03m 52s","accounts_in_message":[],"_revision_number":2},{"id":"9a9c1b5bd6519317520c125727a48485276648f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 13:58:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d0dd31ee35e2cdea81ef1d3fa77ec83cad97ff0a","revisions":{"d412e463bd66900af2db70bb8fdae229922a684b":{"kind":"REWORK","_number":1,"created":"2023-07-28 04:14:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/889911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/11/889911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/11/889911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/11/889911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/11/889911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/11/889911/1"}}},"commit":{"parents":[{"commit":"c3293a495345897d06fc7719049550f7cc313a5d","subject":"Merge \"remove legacy db connection check with proper types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c3293a495345897d06fc7719049550f7cc313a5d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-28 04:10:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-28 04:13:52.000000000","tz":540},"subject":"db: Use ensure_packages to manage backend package","message":"db: Use ensure_packages to manage backend package\n\n... instead of skipping the resource definition in case the same is\nalready defined. This allows us to detect any conflicting definitions\ndefined in multiple manifests.\n\nChange-Id: Id78c6c1553973b57b4b9e7edaf8d029d833f4722\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d412e463bd66900af2db70bb8fdae229922a684b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d412e463bd66900af2db70bb8fdae229922a684b"}]},"branch":"refs/heads/master"},"d0dd31ee35e2cdea81ef1d3fa77ec83cad97ff0a":{"kind":"REWORK","_number":2,"created":"2023-07-28 04:15:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/889911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/11/889911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/11/889911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/11/889911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/11/889911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/11/889911/2"}}},"commit":{"parents":[{"commit":"c3293a495345897d06fc7719049550f7cc313a5d","subject":"Merge \"remove legacy db connection check with proper types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c3293a495345897d06fc7719049550f7cc313a5d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-28 04:10:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-28 04:14:52.000000000","tz":540},"subject":"db: Use ensure_packages to manage backend package","message":"db: Use ensure_packages to manage backend package\n\n... instead of skipping the resource definition in case the same is\nalready defined. This allows us to detect any conflicting definitions\ndefined in multiple manifests.\n\nChange-Id: Id78c6c1553973b57b4b9e7edaf8d029d833f4722\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d0dd31ee35e2cdea81ef1d3fa77ec83cad97ff0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d0dd31ee35e2cdea81ef1d3fa77ec83cad97ff0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
