)]}'
{"id":"openstack%2Fpuppet-oslo~891257","triplet_id":"openstack%2Fpuppet-oslo~master~Ia176b8d881eb9bd1bb5e22b437b4e60ecd6963af","project":"openstack/puppet-oslo","branch":"master","topic":"deprecate-amqp1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-08-17 00:57:24.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"36201":{"account":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"last_update":"2023-08-17 04:01:07.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-08-16 13:34:29.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2023-08-15 14:36:20.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ia176b8d881eb9bd1bb5e22b437b4e60ecd6963af","subject":"Deprecate support for amqp1 driver","status":"MERGED","created":"2023-08-14 01:57:26.000000000","updated":"2023-08-17 04:02:18.000000000","submitted":"2023-08-17 04:01:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891257-deprecate-amqp1","meta_rev_id":"9dc8469a7cdf634359c1e6d9baace644ef021636","_number":891257,"virtual_id_number":891257,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-17 04:01:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-08-16 13:34:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-08-15 14:36:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2023-08-16 13:34:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-14 03:19:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-14 05:43:21.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-08-14 05:55:45.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-08-14 15:58:32.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"5fd2de13adb5beb812ec034870d57f939c3e3a0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-14 01:57:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44384b6a71bbe4bd4e1ee9632cc620db3ad1384c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 03:19:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/741d8f730be64451912dfcb3eef784b1\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/103a473cf3d045a780a1cddfc3f3fd5d : SUCCESS in 2m 21s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43a96ad74b944ced851319f37eb5072a : SUCCESS in 2m 33s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/e34e8a77cd89484fbdd69a5543a169d7 : SUCCESS in 4m 33s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7214468134624145adf93e089597cc02 : SUCCESS in 8m 07s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fc3ae746a754c24929b295fab627da7 : SUCCESS in 6m 15s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/d97d89a45594419ca32b84fe5bfee7ea : SUCCESS in 8m 16s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30c85a10e04d49a4b88deb4328598ecb : SUCCESS in 5m 36s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/192ea378ed3f4383b7a4a2485d88ee81 : FAILURE in 49m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a353a0e196104fd1946cf1d6d9c3a0f3 : SUCCESS in 57m 19s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7754701dc2ec4d24aeb20706f8bc7d5a : SUCCESS in 50m 40s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1a6e88ba265409290d7b90714d69c2d : SUCCESS in 43m 55s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88d7966ce9924bca91cd90d97304b359 : SUCCESS in 44m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f56b5317072f4bc9a9205ee43aa1fb12 : SUCCESS in 1h 11m 15s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5360810b5ff44f2db9b8bc25737ff652 : SUCCESS in 1h 07m 41s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/906215cf508649a58c9cdc70b1b656bc : SUCCESS in 55m 34s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/684daeb2fd6846398ca090223281f0d3 : SUCCESS in 53m 29s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8efcf10c38c4523bc418e54ebf4c5d1 : SUCCESS in 52m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8de1b05a078a4d8c9191d83891391667 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"76d67f70afc6c50b287667ae7ad87835fa122a6e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-08-14 05:43:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"15ff623a562b120ba6facf49fd65079d739f79c7","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-08-14 05:55:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7741ea6d9f3372ae1784fa48ae8c06ce5a980acd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-14 14:23:33.000000000","message":"Topic amqp removed","accounts_in_message":[],"_revision_number":1},{"id":"f511d0b57b41fba4f30f09f20f373c4f2f845809","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-14 14:23:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ddf60bc3697a0652b183a2e6de890b42a4c46791","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-14 14:23:41.000000000","message":"Topic changed from amqp to deprecate-amqp1","accounts_in_message":[],"_revision_number":2},{"id":"e69a31734e221c0ea81b2dda1b6f39de9ded40c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 15:36:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a647242d81c43e1b174ff71f5a1944f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f01ab1c708c40928cb18f838f908a97 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab0f88de1ab6442da97d0b64e50b20f2 : SUCCESS in 2m 27s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/227b423fbb21437f8cc4f14bd9274135 : SUCCESS in 6m 07s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cedc895390274ed0aa5ce6e30470b49b : SUCCESS in 9m 55s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfc7783618a4414196d3fdc52f5e1370 : SUCCESS in 6m 15s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/5cc573bd932c4f50834d4f978f689f26 : SUCCESS in 10m 50s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc3c6aae3f2e4fcea58ba9e97f6c2a05 : SUCCESS in 7m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/138bffeb8f914fd0af3272dd70ffa561 : FAILURE in 50m 20s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93ca4ed79757462dbe7a8b8a210fe736 : SUCCESS in 1h 03m 01s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a822791073984c72ba9bf803f2c41962 : SUCCESS in 42m 11s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcb182f799744f168c821bcb0b4e7309 : SUCCESS in 47m 16s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe7e48292c4a4a918c0754213a95d9f2 : SUCCESS in 37m 48s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/009b9c6d4bc44c2eb4381618dde991ed : SUCCESS in 1h 03m 04s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d60e3cb45fcb4bba9e640e3f6c33fb0b : SUCCESS in 1h 05m 16s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/138c2571f7da42bebbaa6cf2cfe4a5e0 : SUCCESS in 1h 02m 02s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f43b9cfeab84fd0a17b0523693cdb1e : SUCCESS in 50m 39s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e171c2f3fa54af7bf578e10c8b23d87 : SUCCESS in 57m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b74c7d25693485d96eb97da32db8a93 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":2},{"id":"865f6ad97d9afcdabc9fed970f07d9c848b0fcb6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-08-14 15:58:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"93866850a75fdfb0fad941fa3e93550b228c28d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-15 02:31:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a6474ab88c812f68e4a693f5323716b1e9158983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 04:19:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3579be9c2efd4a148b3beb43c3c99969\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/762533bcf3d949af810a37025b3bca8a : SUCCESS in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36b5179b11b64756bdf608f48ecb6013 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c23f9774722e47539ff47602ac1ac4ab : SUCCESS in 5m 29s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd63697559db46e1851c8bf30d6a609a : SUCCESS in 8m 43s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8146fb6ba1b44758a9192eec3128c82c : SUCCESS in 7m 24s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2313680e7914629a6067de2693a4d1d : SUCCESS in 11m 27s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1484d356b508410aa773c867d0fd94d9 : SUCCESS in 7m 35s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/672cf380d5fc4f0e86c7681b5c9d7da4 : SUCCESS in 49m 07s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0644496eb9c34d15a6ec900b4af57d10 : SUCCESS in 1h 00m 55s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be534956bd624e7ca00ca63dc871bb70 : SUCCESS in 41m 41s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ba87692502843bb964aac4a21c8e1d5 : SUCCESS in 45m 09s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/938d3b368f1b4d59b6a1397addfa82eb : SUCCESS in 42m 26s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/db9aa6652764451687e00130ea1bdf1f : SUCCESS in 1h 12m 29s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/26b584eaab7542958cbdd7c54400da12 : SUCCESS in 1h 12m 21s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f67634db1354985be2fc55fe83fefb4 : FAILURE in 1h 08m 18s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb451b1bd1af4e6eab25caff967e7142 : SUCCESS in 55m 53s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d757b4b2892240828a61b54354994452 : SUCCESS in 49m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88d0e6390feb42d2936fa42537b694e7 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":3},{"id":"9abf0cf330e79b6cd105702137a13f6cf91323fa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-15 07:11:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a752bfba2bfd3844ee747a74afcc9c6e375f6e2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 08:11:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faba3ddfc2b44d5fb0c4819adc9cc3aa\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0110f29721c1420ca6940450f8d7df11 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81fc85091d6947f8b9317efb04d77141 : SUCCESS in 2m 40s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/e044a3ce04e24ba3872dfeb6a729fe5c : SUCCESS in 5m 32s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ea6651129914f4c9289514f816a1efc : SUCCESS in 9m 36s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8029b7b909e4066862c996985ab0bc2 : SUCCESS in 5m 51s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/e49b91cd8dae4cb2aa2bd95bc73c8d32 : SUCCESS in 9m 33s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/680dddcec19048eca8bbb58ad8b52031 : SUCCESS in 6m 50s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ff19a8cbeba435a9a00a1177d0d9188 : SUCCESS in 48m 47s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9fe3a0d50c141b6b315375b19dd7cc5 : SUCCESS in 55m 48s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7af6e10d22a74a76b62a1e4b9ad6c9c0 : SUCCESS in 46m 14s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d511ae079b6e40c8a92750c71d046580 : SUCCESS in 42m 33s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9088db49ddf45c291a53e2e0ed73838 : SUCCESS in 45m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2510d4b115f944fe8c00b34f172f0de4 : SUCCESS in 49m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/752000195e5c408f9c1ba40beda0e85b : SUCCESS in 54m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb9a0b7aa822401083c994c3eff1a653 : SUCCESS in 59m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed6a806d22b549fbb9e1b4c395909ba6 : SUCCESS in 44m 44s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0cda02e76ac4ac6aad8e29f167273fe : SUCCESS in 46m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5f6f132fdf2414fb5db75867ab66703 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":3},{"id":"e45cadc841fe4ecec2334137d70748e225254bda","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-08-15 14:36:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8b7bf5e6ddd9e8645e75fd47ab368efce6135b01","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-15 15:41:45.000000000","message":"Topic amqp removed","accounts_in_message":[],"_revision_number":3},{"id":"556afaa25c065b5de6b8b27c30a6df99b0b56484","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-15 15:41:51.000000000","message":"Topic set to deprecate-amqp1","accounts_in_message":[],"_revision_number":3},{"id":"edbc7d8007c5d32b1a8bd112a51b732904eb8c50","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-08-16 13:34:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"769e636558c4064cc1243c076ce8b7cd962f455a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 13:35:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"58ed5f2ba8398e4ee12dc14496a7c7187a674be2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 15:00:18.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e63ff76a34d64bec8b63d2eff7585771\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3c4c0638071437fa7fff23aae3a54fc : SUCCESS in 2m 46s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/408da489b7dc45b48eabe1bbc5b626f6 : SUCCESS in 2m 36s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/fcf5d32c802d4a3bbebeee9400f08114 : SUCCESS in 9m 02s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c24bd55f5104415991d3dfc519d460c : SUCCESS in 6m 16s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/a763fb6c9b71401aa79ff1f69391b539 : SUCCESS in 9m 35s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/44a81cda653b4d4ea1636988909b92bb : SUCCESS in 1h 03m 41s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/be298681690446e9b3d0807ebd93f64d : SUCCESS in 1h 12m 59s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/23e2bd551e4a425ca709c6ff30d59793 : FAILURE in 1h 04m 35s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/35bffbd005654180b8ad85e4cf28231f : SUCCESS in 1h 00m 26s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8145ee35636b4fc698479681b594b0cc : SUCCESS in 1h 00m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7318f6dc81a45db98ba0bf969853dad : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":3},{"id":"c831382965becb9d674252bcbb75a939eddbca12","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-17 00:57:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ccfa089f28841e19a27b12db3b7ffb51c59f600d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 02:20:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/664512dadf5f48ae9d13238430c29e83\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06859f9d9173400baefb18c00ff7ced0 : SUCCESS in 2m 17s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30a2138d888647469d4530876c847605 : SUCCESS in 2m 25s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f2b23bd9d234a3a8f95f34946dd80ad : SUCCESS in 5m 08s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6335b7fcc8146869092147938bb5e06 : SUCCESS in 8m 45s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8505c44c708145308da3997858acc3fd : SUCCESS in 5m 25s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/957e7fec46204f9d86f7d234ae170683 : SUCCESS in 9m 08s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac5dcc8b6171489da208a9a393a60405 : SUCCESS in 5m 49s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/679a7e1c47144833a27473738980ea70 : SUCCESS in 35m 35s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/325e003ce4914e4d8b8f6af348c09884 : SUCCESS in 49m 54s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f482dfc650de435d9ca97cadbd009503 : SUCCESS in 38m 54s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f569d019365f4e469889e4fd4a3db51b : SUCCESS in 35m 49s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90a6f9bc27a542a29cd6e759ac206e02 : SUCCESS in 36m 43s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c62d22a35a548f8a44594e31b67c9dc : SUCCESS in 1h 03m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f9627bcf6534bfe9d1c4dd26d51fc97 : SUCCESS in 1h 17m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ca3f5355dc247d0b9f5f8e4d3fb39ec : SUCCESS in 55m 43s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/03835c8a3cb344f694cbcbb8382d3378 : SUCCESS in 45m 35s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f690b5d887e479b956f04063ee3bc42 : SUCCESS in 44m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e9e0407a6284e76be8ea49687ecffa9 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"deb79bf5ee38d471ae60e5f8cee45834dca72f74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 02:21:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f469a5d66be8b98af88ddd6ed7d353156f91b53f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 04:01:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3391cc2270f8451281d376be3fb9dad0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/271dff572cd64a60900fd558936b055e : SUCCESS in 3m 30s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da01f9bcc9284498add5aee7802e23a6 : SUCCESS in 2m 43s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3e5fc537fe14135a3280631b7ca1364 : SUCCESS in 10m 28s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41a510376b0548b3bf50e8fc2b31666a : SUCCESS in 6m 40s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2d8ff0ea8e542eb86dfa9f933cfcbce : SUCCESS in 11m 32s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/758424d8f9ad491f884fbbb01abb62fc : SUCCESS in 1h 08m 18s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3d2d3e647684ca5b31b3ce2eef31d20 : SUCCESS in 1h 16m 30s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5116efb10bd4029bd0c5658a1bd234b : SUCCESS in 1h 02m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a314959e8bde40c3b410d8aa7564045f : SUCCESS in 1h 02m 22s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc541d84f17d401c8511b2194b2643ef : SUCCESS in 59m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df44e9feaf714347a9f47fdb092a5956 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":3},{"id":"3f100157258d51f8baf32950699b19fa1f309a85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 04:01:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9dc8469a7cdf634359c1e6d9baace644ef021636","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 04:02:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ba9a21afefa4cfd82718da45f808a2e\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d86103b2977d4e179afab712aff92fc7 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e174113210699b433dbae4c2c5a68bc3d269b5c3","revisions":{"66cd0e983db10137525ae72554c747099006c2dc":{"kind":"REWORK","_number":1,"created":"2023-08-14 01:57:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/891257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/57/891257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/57/891257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/57/891257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/57/891257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/57/891257/1"}}},"commit":{"parents":[{"commit":"1eedd9bcc7ee169ff405043f59627fa093980ee0","subject":"Add .pdkignore to exclude some files from tar ball","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/1eedd9bcc7ee169ff405043f59627fa093980ee0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-14 01:39:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-14 01:56:20.000000000","tz":540},"subject":"Deprecate support for amqp driver","message":"Deprecate support for amqp driver\n\nThe amqp driver has been deprecated in oslo.messaging because of\nlack of qpid_dispatch module. This deprecates support for the driver.\n\nBecause deprecating all parameters for amqp driver needs large changes\nin multiple modules, this change introduces the logic to detect usage\nof amqp1 driver and shows the warning instead.\n\nDepends-on: https://review.opendev.org/861070/\nChange-Id: Ia176b8d881eb9bd1bb5e22b437b4e60ecd6963af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/66cd0e983db10137525ae72554c747099006c2dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/66cd0e983db10137525ae72554c747099006c2dc"}]},"branch":"refs/heads/master"},"9676c66dc382b7a508ea3def9bc379d0729d80cb":{"kind":"REWORK","_number":2,"created":"2023-08-14 14:23:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/891257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/57/891257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/57/891257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/57/891257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/57/891257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/57/891257/2"}}},"commit":{"parents":[{"commit":"1eedd9bcc7ee169ff405043f59627fa093980ee0","subject":"Add .pdkignore to exclude some files from tar ball","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/1eedd9bcc7ee169ff405043f59627fa093980ee0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-14 01:39:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-14 14:23:27.000000000","tz":540},"subject":"Deprecate support for amqp1 driver","message":"Deprecate support for amqp1 driver\n\nThe amqp1 driver has been deprecated in oslo.messaging because of\nlack of qpid_dispatch module. This deprecates support for the driver.\n\nBecause deprecating all parameters for amqp1 driver needs large changes\nin multiple modules, this change introduces the logic to detect usage\nof amqp1 driver and shows the warning instead.\n\nDepends-on: https://review.opendev.org/861070/\nChange-Id: Ia176b8d881eb9bd1bb5e22b437b4e60ecd6963af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9676c66dc382b7a508ea3def9bc379d0729d80cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9676c66dc382b7a508ea3def9bc379d0729d80cb"}]},"branch":"refs/heads/master"},"e174113210699b433dbae4c2c5a68bc3d269b5c3":{"kind":"REWORK","_number":3,"created":"2023-08-15 02:31:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/891257/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/57/891257/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/57/891257/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/57/891257/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/57/891257/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/57/891257/3"}}},"commit":{"parents":[{"commit":"1eedd9bcc7ee169ff405043f59627fa093980ee0","subject":"Add .pdkignore to exclude some files from tar ball","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/1eedd9bcc7ee169ff405043f59627fa093980ee0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-14 01:39:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-15 02:30:54.000000000","tz":540},"subject":"Deprecate support for amqp1 driver","message":"Deprecate support for amqp1 driver\n\nThe amqp1 driver has been deprecated in oslo.messaging because of\nlack of qpid_dispatch module. This deprecates support for the driver.\n\nBecause deprecating all parameters for amqp1 driver needs large changes\nin multiple modules, this change introduces the logic to detect usage\nof amqp1 driver and shows the warning instead.\n\nDepends-on: https://review.opendev.org/861070/\nChange-Id: Ia176b8d881eb9bd1bb5e22b437b4e60ecd6963af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/e174113210699b433dbae4c2c5a68bc3d269b5c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/e174113210699b433dbae4c2c5a68bc3d269b5c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
