)]}'
{"id":"openstack%2Fpuppet-oslo~928497","triplet_id":"openstack%2Fpuppet-oslo~master~I779ae835d29b24605c1651048a46b01c0a2587b2","project":"openstack/puppet-oslo","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-09-10 10:13:37.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I779ae835d29b24605c1651048a46b01c0a2587b2","subject":"Add support for [oslo_messaging_rabbit] rabbit_stream_fanout","status":"MERGED","created":"2024-09-09 05:16:36.000000000","updated":"2024-09-10 13:13:54.000000000","submitted":"2024-09-10 13:12:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928497","meta_rev_id":"6a5948d1f680a67c74bda5cc0eeb2241a0a3ebc2","_number":928497,"virtual_id_number":928497,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-10 13:12:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2024-09-09 21:39:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2024-09-10 07:38:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2024-09-10 07:38:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-09 06:34:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-09 21:39:57.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-09-10 07:38:49.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"8168e48cda608b27ca4f756b0440ca8c7ecf4ad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-09 05:16:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b230f22f0652db43316a1e3259381a78655aae5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 06:34:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da6a1f111081473ab61d364eca7eaa71\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87c12e09b53549f08ca438d7c29810c5 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f94b7d46d3641a8a704f0512680e207 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/21cde7f2d1c146f189689bec40f234d0 : SUCCESS in 5m 55s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/0798febbc01040a2af7ac0f47fe4ce11 : SUCCESS in 7m 38s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0bcb4093f60d47f8b83553fa73e82870 : SUCCESS in 5m 00s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/11c49e7a9e5c43629d98f80bb04fc5c9 : SUCCESS in 5m 47s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0aed7ad3e92f4977bafa5c130c8b7687 : SUCCESS in 4m 51s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/328b47425cff4e99aa9600d233d683fc : SUCCESS in 44m 27s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/decd68c4882c47ceaa3d1db3ce042aea : SUCCESS in 59m 12s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24af3b10f44b4e64a742481ebe728350 : SUCCESS in 42m 13s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2967a46261464d32a4a302600dbc1cab : SUCCESS in 40m 38s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ea329bc431e4191aae413ec870ebe29 : SUCCESS in 36m 55s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8956d2070603469e975cda7dba45ae9c : SUCCESS in 56m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/89e881b834a74dddbe4efe417745581e : SUCCESS in 1h 10m 28s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b573a5d8ba174c7785cc8a65ee3859f1 : SUCCESS in 1h 05m 49s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c16832ba02cf46678d590d4e5f8ef80d : SUCCESS in 57m 22s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/47adbf7959d14b8296c71aa082c58188 : SUCCESS in 53m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f655b152a31403f9ac13eb0eded9653 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"fc420aa9a71be497e7c45c5003c26d1968dd330f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-09-09 21:39:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3bc1a018d896a41787c42d97ee94eb2ab5fde8d6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-09-10 07:38:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"99629730867ba692d4f04dffd3277c788b262f6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 07:39:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f80ac1e76eb701e9902a33f4b8020ea2f7c60c0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 08:55:30.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/928496 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"7e85933ba01ae3e3bbfe2001edcac04845db1576","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-10 10:13:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d942d2fe37275cec19e64f2c42bc5cf2c6bdc2d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 11:36:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd2d992698d94e9aa9958b64c025260e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c86a63439ee43e9a53831345a833df1 : SUCCESS in 2m 31s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b93d5d686744182808d6747a0c7ed5d : SUCCESS in 3m 29s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/12bb0d2a18e449a5aa7f3ddbe66c43d5 : SUCCESS in 4m 37s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0c5a6d691034a468e6f21b411c29ac4 : SUCCESS in 5m 07s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b476d88ba5e49ac820bb0bf3289cfba : SUCCESS in 4m 43s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/49cf24df7b3e41e6ab8b88ba66011165 : SUCCESS in 5m 01s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4b25a289b064b9989ef9bb143f64295 : SUCCESS in 5m 01s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1eadf471cf414a04a67ac0275b4b65dd : SUCCESS in 43m 21s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4079add01805471d913b51b9b9dd6d75 : SUCCESS in 57m 34s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87e8282180814323adc9e0f84356d365 : SUCCESS in 42m 30s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9367ce96f9d04fa48e79d317f0d7d9b2 : SUCCESS in 42m 20s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f258c8d0aabc4cd19dcbb0c4f771e20b : FAILURE in 45m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/33a3e40026014bb389f4063f96d6b496 : SUCCESS in 1h 02m 38s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b8ad57970404d8fac7ea065e4545efe : SUCCESS in 1h 17m 26s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/10c7a968aee4420ab43512f8be2788a1 : SUCCESS in 57m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/56ce15279ca4467b94fb6de2fdd0ef0c : SUCCESS in 33m 52s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a6425e9e1a848c890aa29e7cb16555d : SUCCESS in 57m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/049c8bd5f10a4d3390a3c900586a4fea : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a109ceff5c0f41d93ff5034443ecc9c447bc43d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 11:44:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d323fe0c6b6a9d15d3cc228853db15daf5d656b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 13:12:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95843613c4fa4448935c10531a4dd87b\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ea6c26d9d8f40f68632e36350be0899 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b96ad8718384a7dba59d2c3cd022b90 : SUCCESS in 2m 19s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/4cf11c6d35fd405f82f2cd6c1605a4e2 : SUCCESS in 6m 43s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3de10fa664a24f5caf814463439e957f : SUCCESS in 3m 35s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fadce4f1994416cbebb20b6ac50d3d0 : SUCCESS in 51m 52s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c17a77ce1c9e4662a723e3aa35a35fe4 : SUCCESS in 1h 04m 47s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/22166e9c41274861976344c606eca8d5 : SUCCESS in 37m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e92593dd14e64aa1852abd37660001c8 : SUCCESS in 57m 32s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e73b61167d9a4238a595614e957b119d : SUCCESS in 53m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/caad4ba9afff408fa3edc1d4790e74a2 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"00a5014dce2101896941d51a03ff862a6eebf77e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 13:12:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6a5948d1f680a67c74bda5cc0eeb2241a0a3ebc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 13:13:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a666d347ba8b439692817b5b9e568904\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a5e93f4a079459f91414e52ab2cde23 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac4e5ee56f94ac9ae6cd4594991df06b9032d820","revisions":{"ac4e5ee56f94ac9ae6cd4594991df06b9032d820":{"kind":"REWORK","_number":1,"created":"2024-09-09 05:16:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/928497/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/97/928497/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/97/928497/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/97/928497/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/97/928497/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/97/928497/1"}}},"commit":{"parents":[{"commit":"584440d30cd5c5a5ff7fd5bb241bb72888ec2198","subject":"rabbitmq: Support queue manager options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/584440d30cd5c5a5ff7fd5bb241bb72888ec2198"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-09 05:15:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-09 05:16:25.000000000","tz":540},"subject":"Add support for [oslo_messaging_rabbit] rabbit_stream_fanout","message":"Add support for [oslo_messaging_rabbit] rabbit_stream_fanout\n\nChange-Id: I779ae835d29b24605c1651048a46b01c0a2587b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/ac4e5ee56f94ac9ae6cd4594991df06b9032d820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/ac4e5ee56f94ac9ae6cd4594991df06b9032d820"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
