)]}'
{"id":"openstack%2Fpuppet-oslo~932062","triplet_id":"openstack%2Fpuppet-oslo~master~I22bf39917fbfa9b67e36a3bf46f96faae9b0011c","project":"openstack/puppet-oslo","branch":"master","topic":"oslo-limit-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-11 16:44:56.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2024-10-11 15:09:42.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I22bf39917fbfa9b67e36a3bf46f96faae9b0011c","subject":"limit: Support new options to query endpoint","status":"MERGED","created":"2024-10-10 12:29:47.000000000","updated":"2024-10-11 16:46:01.000000000","submitted":"2024-10-11 16:44:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932062-oslo-limit-opts","meta_rev_id":"c437a89624b5cc5e89b6eccedcd64917bf6a0c83","_number":932062,"virtual_id_number":932062,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-11 16:44:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2024-10-11 15:09:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2024-10-11 06:42:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2024-10-11 15:09:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-10 14:24:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-10 14:50:49.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-10-11 06:42:44.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"b13415882cd9d52b8317e0b5466f316f1dbadf08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-10 12:29:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c15021de07266d67f76eecb24138c6684cef07b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-10 12:44:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e3b9e2eddb26e31a0e11a3750834700cb27dc649","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-10 13:01:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"71ed982e170095b5591d71c48d18482665567467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 14:24:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/98ac4c87b77b49f0a13d538ae69be552\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d557c20a2cca49818ec984426977f433 : SUCCESS in 2m 26s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98c0cc5fa45c4392a409ec44e9a63315 : FAILURE in 2m 41s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bfd553202d64e2eb564fc5cf2f1d460 : FAILURE in 5m 26s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/c277541c739049fba4075487b81f5b72 : FAILURE in 8m 16s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0512f389ffd45c1855443c8828d3d25 : FAILURE in 5m 03s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/df1d5dd2ef324a3bb81c8ca57cc5002e : FAILURE in 8m 20s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e97b2b6be0a45cf86cd1e4621245a1f : FAILURE in 4m 39s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81d157bc3d4c4541a3a048c8b29a4193 : SUCCESS in 44m 18s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ea42e95694a4b44a31d2673e6dcd62b : SUCCESS in 1h 05m 00s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fa569b91c1744aeb03b7a0f0430fa54 : SUCCESS in 42m 02s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f789cea12ab84c6a8c22fd6d4c7ab3eb : SUCCESS in 42m 56s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c71fce14bd934632a03442fb55973fc9 : SUCCESS in 41m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0dc873d74b34266b21161890ccecf33 : SUCCESS in 55m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f45e9ffefdac42f09197f9ba9cf74797 : SUCCESS in 1h 16m 46s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c041217620d45fbbd488eea7e4ea425 : SUCCESS in 57m 23s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/38083e86279d4318a34178a9dcf5e919 : SUCCESS in 53m 46s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6544fbfd69784f0f8b3c305e4d4ff1d8 : SUCCESS in 52m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7cb5ddd9b844907931e90507e0a12de : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":3},{"id":"89c7587d3177cd08d366b3274cd0ddd093460085","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-10 14:31:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"80cdaa08a5284850cfed7f52197943336742a53e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-10-10 14:50:49.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"52b03362281fad433e3dbd7efe3c7c530de0020c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-10 15:17:06.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"9b220f83121a8a73371b19400f50889a02b6545e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 16:36:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/585e160f19a14075b89e04c89e236a40\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83ba998f7cea47edbf252895859700f2 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b2a99b9a6c24ec5ab06f52287a8f3ef : SUCCESS in 3m 18s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d273a0e2f5ed4ac48c22cde9aed341ad : SUCCESS in 5m 18s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf4d7fae2d594dc4802ec5b83cdaec9f : SUCCESS in 8m 02s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6bc34944f86a49f5b170c3e181cc1559 : SUCCESS in 6m 19s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/5faa6d54ede647d59afed11c0b9c266c : SUCCESS in 8m 17s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f5c5e609ce84071ac2dc3eede5eab33 : SUCCESS in 5m 02s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32e9da9de9c74e268a524e36a9a8ed50 : SUCCESS in 49m 22s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c74edd67c0c34113bd6ad74866c4186d : SUCCESS in 56m 28s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7df59a3470c04e6aa721222edd8228a5 : SUCCESS in 45m 49s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08e59d0b2b4b42d7a8a23b54c48bbdde : SUCCESS in 40m 18s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99e4d58cda3444b8b31d65fc6ce22e0b : SUCCESS in 42m 33s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c9afb76b05c45f1a4fafc6bb009fc2c : SUCCESS in 50m 04s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f76df1757ef41a1aa249ccd9d0e2570 : SUCCESS in 1h 11m 37s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba8334863a7941ab88cad799d97004ad : SUCCESS in 56m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c46d7ea3f1694a49895cf52c0b0a0bfe : SUCCESS in 55m 38s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/950bcd1807584bcc9c14c6c8eb14d849 : SUCCESS in 53m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6343e10cf5cc479e94ded003bdca0eb4 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":5},{"id":"1cc4cc42148cfcf1e9bd9a1d2eb794051100fb0a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-11 06:42:44.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"bd2c4dd840aacecb6c3cc458f06d06f959720396","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-11 14:40:29.000000000","message":"Topic set to oslo-limit-opts","accounts_in_message":[],"_revision_number":5},{"id":"1818f7f252a63306b6d891dd3618a246aa0ab1b0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-10-11 15:09:42.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"99c592d6a148482c8ffc709d9f01b1d317a01228","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 15:10:32.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9b3692f6634810638a4bea6aca202a759b1a3e32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 16:44:56.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbc8a43631df4a688a604911c52d937b\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac0357dca981497f9395b5d67447661b : SUCCESS in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7732e960283a409aa4280a23e14602d4 : SUCCESS in 2m 08s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9bad41100274141be6c3659372b2336 : SUCCESS in 7m 57s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a574763e73148a6bbbb387470bdc426 : SUCCESS in 4m 56s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8d197f4919b4f42a7008815de7a3e56 : SUCCESS in 1h 02m 07s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/973dc7a281f14b668142853c612ca131 : SUCCESS in 1h 28m 53s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dffea2251b13409b82d62d12439af6c0 : SUCCESS in 56m 24s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cf81b029db44688aeee8e63da49df38 : SUCCESS in 56m 43s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6f3acb606a24baa9537ae5a75b76d35 : SUCCESS in 54m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae737ca986d540b2b837410c8c3d434d : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":5},{"id":"d9e50dc54212746159d2c10b4c18c9bf6bc806fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 16:44:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"c437a89624b5cc5e89b6eccedcd64917bf6a0c83","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 16:46:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9c1ffad367349e6b2bf935150efa990\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f97e4bf850724bcaa36aa20c494516fc : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"da9434eefd6cba0fdea861f76adf77c72e189692","revisions":{"3977509987dc1d62d3c06049aa6100689f241cfb":{"kind":"REWORK","_number":1,"created":"2024-10-10 12:29:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/932062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/62/932062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/1"}}},"commit":{"parents":[{"commit":"ebdef74a822d77e4709945c1508fdba83ef65804","subject":"ssl: Reject if only one of cert_file and key_file is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/ebdef74a822d77e4709945c1508fdba83ef65804"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-10 12:28:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-10 12:29:36.000000000","tz":540},"subject":"limit: Support new options to query endpoint","message":"limit: Support new options to query endpoint\n\n... and also adds a few more useful default values.\n\nDepends-on: https://review.opendev.org/914783\nChange-Id: I22bf39917fbfa9b67e36a3bf46f96faae9b0011c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3977509987dc1d62d3c06049aa6100689f241cfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3977509987dc1d62d3c06049aa6100689f241cfb"}]},"branch":"refs/heads/master"},"17173277afdd6a81952f8541a7f096e53f14cef7":{"kind":"REWORK","_number":2,"created":"2024-10-10 12:44:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/932062/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/62/932062/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/2"}}},"commit":{"parents":[{"commit":"ebdef74a822d77e4709945c1508fdba83ef65804","subject":"ssl: Reject if only one of cert_file and key_file is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/ebdef74a822d77e4709945c1508fdba83ef65804"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-10 12:28:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-10 12:43:57.000000000","tz":540},"subject":"limit: Support new options to query endpoint","message":"limit: Support new options to query endpoint\n\n... and also adds a few more useful default values.\n\nDepends-on: https://review.opendev.org/914783\nChange-Id: I22bf39917fbfa9b67e36a3bf46f96faae9b0011c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/17173277afdd6a81952f8541a7f096e53f14cef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/17173277afdd6a81952f8541a7f096e53f14cef7"}]},"branch":"refs/heads/master"},"b85cc90715579be01302558a9e2d2c5bb0c06cc0":{"kind":"REWORK","_number":3,"created":"2024-10-10 13:01:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/932062/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/62/932062/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/3"}}},"commit":{"parents":[{"commit":"ebdef74a822d77e4709945c1508fdba83ef65804","subject":"ssl: Reject if only one of cert_file and key_file is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/ebdef74a822d77e4709945c1508fdba83ef65804"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-10 12:28:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-10 13:01:25.000000000","tz":540},"subject":"limit: Support new options to query endpoint","message":"limit: Support new options to query endpoint\n\n... and also adds a few more useful default values.\n\nDepends-on: https://review.opendev.org/914783\nChange-Id: I22bf39917fbfa9b67e36a3bf46f96faae9b0011c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b85cc90715579be01302558a9e2d2c5bb0c06cc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b85cc90715579be01302558a9e2d2c5bb0c06cc0"}]},"branch":"refs/heads/master"},"89ea84efa8eb2df6786d210a46e7588a0b9b7093":{"kind":"REWORK","_number":4,"created":"2024-10-10 14:31:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/932062/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/62/932062/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/4"}}},"commit":{"parents":[{"commit":"ebdef74a822d77e4709945c1508fdba83ef65804","subject":"ssl: Reject if only one of cert_file and key_file is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/ebdef74a822d77e4709945c1508fdba83ef65804"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-10 12:28:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-10 14:31:12.000000000","tz":540},"subject":"limit: Support new options to query endpoint","message":"limit: Support new options to query endpoint\n\n... and also adds a few more useful default values.\n\nDepends-on: https://review.opendev.org/914783\nChange-Id: I22bf39917fbfa9b67e36a3bf46f96faae9b0011c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/89ea84efa8eb2df6786d210a46e7588a0b9b7093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/89ea84efa8eb2df6786d210a46e7588a0b9b7093"}]},"branch":"refs/heads/master"},"da9434eefd6cba0fdea861f76adf77c72e189692":{"kind":"REWORK","_number":5,"created":"2024-10-10 15:17:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/932062/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/62/932062/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/62/932062/5"}}},"commit":{"parents":[{"commit":"ebdef74a822d77e4709945c1508fdba83ef65804","subject":"ssl: Reject if only one of cert_file and key_file is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/ebdef74a822d77e4709945c1508fdba83ef65804"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-10 12:28:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-10 15:16:54.000000000","tz":540},"subject":"limit: Support new options to query endpoint","message":"limit: Support new options to query endpoint\n\n... and also adds a few more useful default values.\n\nDepends-on: https://review.opendev.org/914783\nChange-Id: I22bf39917fbfa9b67e36a3bf46f96faae9b0011c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/da9434eefd6cba0fdea861f76adf77c72e189692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/da9434eefd6cba0fdea861f76adf77c72e189692"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
