)]}'
{"id":"openstack%2Fpuppet-oslo~944132","triplet_id":"openstack%2Fpuppet-oslo~master~I7069781a6f97beb626a8e79a03de9f4182e8372a","project":"openstack/puppet-oslo","branch":"master","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2025-03-13 07:35:04.000000000","reason":"\u003cGERRIT_ACCOUNT_6476\u003e replied on the change","reason_account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}}},"hashtags":[],"change_id":"I7069781a6f97beb626a8e79a03de9f4182e8372a","subject":"Add an amqp_auto_delete param","status":"MERGED","created":"2025-03-12 16:04:50.000000000","updated":"2025-03-13 10:00:10.000000000","submitted":"2025-03-13 09:59:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944132","meta_rev_id":"b8d419f2837ba7d63f0c32e2e27e4ccb2158848d","_number":944132,"virtual_id_number":944132,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-13 09:59:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-03-12 16:18:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-03-12 19:49:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-03-12 19:49:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-12 16:18:52.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-03-12 17:23:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-12 19:49:58.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"fce4e61f1795cc575243239656c711646d689986","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-03-12 16:04:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e14e7faf8057ad4bdb29019aead1f36725c10851","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-03-12 16:18:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1cd8bb45b4e439c5662fa9da5bce57cf6d4d470d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 17:23:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae30033aa9e2476889340f617d5bb84f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/197759bdae72407c91a48394e704618c : SUCCESS in 3m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41f36c51c61847b2a1d3613938a6f980 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/67fd55e4fbd34746adadaa9579d8355f : SUCCESS in 5m 02s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/127c97663bb54d89b8ba3dd115e83722 : SUCCESS in 7m 10s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b0fbc018b114dc8bb6f28f3ef950d35 : SUCCESS in 4m 38s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6c62e555af445f8abbda85e4bba3a33 : SUCCESS in 6m 59s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/718602a4ff394f24a482ffd578e38f6b : SUCCESS in 3m 02s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4727f824bb74d9d97725a367918e14f : SUCCESS in 25m 32s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cf5266820234ac4a3853448deedf99c : SUCCESS in 55m 26s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a66df04c778a49fba99899c4dbde1286 : FAILURE in 32m 16s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54fa4b17af8c4d9db6be18825bba7bb5 : SUCCESS in 38m 11s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c8c8f0a18c6463f96ba8c78ee4dfaab : SUCCESS in 44m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a37617ae3e3e44edb6fb167b42a2a037 : SUCCESS in 1h 02m 35s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1a841c2ec8a4c58a7748730509a2a89 : SUCCESS in 1h 06m 58s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca632468de7643b3857dd1c51266f379 : SUCCESS in 1h 00m 34s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd859026f2204cd093c46e48fb4e899f : SUCCESS in 49m 36s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/63be27cd3fae42a7aa4659514fd4aaec : SUCCESS in 44m 14s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4fa77371a214c55b1589815a5fa297f : SUCCESS in 1h 03m 53s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5726ea56c534ad08d3e75d2df0413a4 : SUCCESS in 1h 17m 32s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1217de9b35294f14b2a67d89d4f1d15f : SUCCESS in 53m 39s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4de5e8ecac224ef1a958e4e85c161b2d : SUCCESS in 43m 59s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b014f66cc9784f4a968696a19e134050 : SUCCESS in 34m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52cbe751974441efa8c643b345c226d8 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e9d8e0d0249a40f4e02dd501a52133cc6a890678","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-12 19:49:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"99568e356da841821f729c07dc49a98f1e11a85b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 19:50:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"148ded3eb7521ef76b0b28cf1c06e381ea995ddc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 21:01:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bf51b995a4b476f81dbf6b10dcdc9f8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29fa4e28e58348a4a48a7cb12e248384 : SUCCESS in 2m 25s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4027ac096c564596aa150464a14c5a1d : SUCCESS in 2m 56s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/951f9bddbf2f431fb1a319db42bc4931 : SUCCESS in 8m 10s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47b6c32c76584121ba19ab57873a144a : POST_FAILURE in 3m 12s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8134882ed6f4d94a4b7ac76fbc6ab3e : POST_FAILURE in 9m 18s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/53ec8b8a79724697812315edb3334a2b : SUCCESS in 1h 02m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/66cdbb9e65d9475fb4cf291c50ab839a : SUCCESS in 44m 24s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/584ff95814b1413c893affc289a8b6b0 : SUCCESS in 26m 34s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/52b2506094bc47faa2eff79b165a5640 : SUCCESS in 52m 37s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c97acb7fc89f42deb2b2845ffb75fd81 : POST_FAILURE in 55m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cad9476d2c1343049c775e26fe036ed5 : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":1},{"id":"54c9460737e7ad1a611a4d990838ec6527a11ebc","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-03-13 07:35:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f564346a73dbd0f909a836b35583e9f2d5518bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 09:17:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87aff5af20cd4256816c40a6e9216e3c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b263d981cd03461fa45b86740b23d311 : SUCCESS in 5m 23s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9316014682848b987273752455a7a26 : SUCCESS in 2m 34s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/69c7a334fd0e4a75901f9cfd71b16b40 : SUCCESS in 5m 52s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f62c4e36ed34724aad0cc8717cf6fad : SUCCESS in 7m 41s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fb96f1dc12245f1b5679f7aba5d2048 : SUCCESS in 2m 29s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fa805ba82ff4e2b92fe226a64c9bed6 : SUCCESS in 7m 59s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b60781da99b94eb2a5e0ee5d7d7429e8 : SUCCESS in 3m 52s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3145a32365f45f3ac32811f89fb5aff : SUCCESS in 45m 37s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74c4864b2759490c8bc7788ede72bdd9 : SUCCESS in 55m 09s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/795a3aa23d544188b8f67094844f5322 : SUCCESS in 48m 06s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/671625fc565042e7aa7d201b6ca19681 : SUCCESS in 24m 14s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9ea271c3b5448749077f5fe7856c8b1 : SUCCESS in 40m 52s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbe05edbabcf4c13883b2507640418f8 : SUCCESS in 28m 36s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/38575cb56b3f4a0bb7254d0a2bba52af : SUCCESS in 36m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac3c09fb44fb4e3fbb4745300f598180 : SUCCESS in 50m 48s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/39e41f5e28ca41af89835a3774821dbd : SUCCESS in 31m 37s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e149887f6cc2489391d4f46aedeedf54 : SUCCESS in 58m 31s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5aaeb7520684ec5800c1c3549346d70 : SUCCESS in 1h 03m 50s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/62f9783924624c2d89449e5bc8c0863e : SUCCESS in 1h 11m 41s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb901a3883074a948c74e337108576b1 : SUCCESS in 57m 45s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9976345d7f8e4cde924178941a0bcefc : SUCCESS in 49m 20s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b3a45074684438c9a72fce7155d9802 : SUCCESS in 32m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87994bb9371243fd8402150890db76cd : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":1},{"id":"92c000c7d4b16ca778d349aa150946602fa6a19a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 09:18:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ece39dfb5603906626a87519d14eded37b87baf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 09:59:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b95999a43b049ffbd346a4a278ccdda\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e96c7d33b294214b9555832074a798e : SUCCESS in 2m 20s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/965f242294e249649c89d8bcb6bf3495 : SUCCESS in 2m 18s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a79910fabe746e8952030508b5a4a6e : SUCCESS in 6m 17s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a62305ae8dd4f7d89b80ad7f758d859 : SUCCESS in 4m 03s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b8c5e59613b435e8afd56784fe31e2e : SUCCESS in 6m 11s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/31e14b8a37c24d12a5d5c45297f02e5f : SUCCESS in 30m 12s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/791ba6cefbe7428fa9c37566e90d5707 : SUCCESS in 40m 03s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9c5b590725549158b2ba739683579bf : SUCCESS in 25m 29s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7e1f1cd14b444978724af1d3cc63dc9 : SUCCESS in 32m 01s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c129eb2dd5ed43c380a816fd6918739f : SUCCESS in 31m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c14999d738db45de93273603f3043c99 : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b6b632e872d85c8a1beac6dde260703a4ebbe785","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 09:59:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b8d419f2837ba7d63f0c32e2e27e4ccb2158848d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 10:00:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6040ff4404544d5fa4f059fabad49390\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd51d57365964c348f8ab579b873999b : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"94b10397b42c2425a5a3d457e034ebf37a41bf6d","revisions":{"94b10397b42c2425a5a3d457e034ebf37a41bf6d":{"kind":"REWORK","_number":1,"created":"2025-03-12 16:04:50.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/32/944132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/32/944132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/32/944132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/32/944132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/32/944132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/32/944132/1"}}},"commit":{"parents":[{"commit":"06a9b12199ed03842bb935749dedf4fc8e20d043","subject":"Deprecate heartbeat_in_pthread parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/06a9b12199ed03842bb935749dedf4fc8e20d043"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-03-12 16:04:36.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-03-12 16:04:36.000000000","tz":60},"subject":"Add an amqp_auto_delete param","message":"Add an amqp_auto_delete param\n\nChange-Id: I7069781a6f97beb626a8e79a03de9f4182e8372a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/94b10397b42c2425a5a3d457e034ebf37a41bf6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/94b10397b42c2425a5a3d457e034ebf37a41bf6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
