)]}'
{"id":"openstack%2Fpuppet-oslo~963186","triplet_id":"openstack%2Fpuppet-oslo~master~I7359f3826c61b058a5f0261e2229b07c31f32e88","project":"openstack/puppet-oslo","branch":"master","topic":"dev_add_db_connection_parameters","attention_set":{},"removed_from_attention_set":{"7888":{"account":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"last_update":"2026-05-11 21:33:26.000000000","reason":"\u003cGERRIT_ACCOUNT_7888\u003e replied on the change","reason_account":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-25 15:45:32.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2026-04-13 21:07:33.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I7359f3826c61b058a5f0261e2229b07c31f32e88","subject":"add posibility to specify db connection parameters","status":"MERGED","created":"2025-10-06 14:03:45.000000000","updated":"2026-05-12 00:24:56.000000000","submitted":"2026-05-12 00:23:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963186-dev_add_db_connection_parameters","meta_rev_id":"1de289da607d0fcb9f0f2d938c7c4b6084d0d9af","_number":963186,"virtual_id_number":963186,"owner":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-12 00:23:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-04-25 15:45:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-04-13 21:07:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-04-25 15:45:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-06 14:52:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-10 18:14:05.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2026-03-25 20:16:46.000000000","updated_by":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"485f5ac36f1fe190f436e5cc5f705e8b3613b055","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2025-10-06 14:03:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87f19af668ff967ef13e1ed3a95d2fe34c3962e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2025-10-06 14:35:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ab1f4318161e61259e30757c2c9b142254f16038","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 14:52:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a24116052dd8473ba92c4ea98cb5b3ec\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9d7e678a71d4c5891032005f1fcdfc3 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1cc67bc549b43c9923b43b655bb6f06 : SUCCESS in 1m 50s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/81319c5733b04a5aaf057a88beea42a2 : FAILURE in 3m 47s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2462d56d31ac4dadaa460104f7b2ba87 : FAILURE in 4m 28s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/723a1ba2c5f84b4ba6281fff18adeccc : FAILURE in 4m 27s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0db459069f54b0f9761b302b7f8412a : FAILURE in 4m 20s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c627cee8edfa456c810427e6c644e1c4 : FAILURE in 4m 39s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d61ba0d13d44f7fb8d9decbfe5203d7 : FAILURE in 5m 29s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8b6fb0bfe4d482d9650324ff85eb511 : FAILURE in 5m 55s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b42a1da5014b40b1b189a6201bff5967 : FAILURE in 5m 32s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/970abaeab7c84a0e9e4b5aa9b5cbb523 : FAILURE in 9m 39s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b9352bd6d62499d8f6d09159978e558 : FAILURE in 8m 30s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6cd521b8b164dcaa0cac6662f1845ea : FAILURE in 5m 18s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4cd513d4a4545bdb5f3dd81611c2860 : FAILURE in 5m 15s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f21f7759481435ebf97c59506454779 : FAILURE in 6m 11s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6bda3762e98945ff863c7f31eaa20970 : FAILURE in 10m 20s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aaf949dd9fb84a13851b556aa01aa0a4 : FAILURE in 10m 36s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8730b108b9f54c4a974ed5c79eee6952 : FAILURE in 10m 04s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3418c69af47b46b3b0ec0f3901135a73 : FAILURE in 5m 19s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3eb41e6eb764f44a62155c499b74d2f : FAILURE in 9m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/160d1922456d40ab89f1c200283847c9 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"0aff09634aaaba2913d68ce773931143c63d6ec0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2025-10-06 15:17:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"24f06f64694dc52382cfeff0bcc90ffd43389a15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 16:49:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d653fc5f9b604c1baeefa85d69169b90\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/272b99fbe128481293ebe7894ed7782b : SUCCESS in 2m 43s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c9d5c5b6bcd4481a054de5e4902506f : SUCCESS in 1m 45s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/d11e71e7a05b467b81be5bb634175e3a : SUCCESS in 3m 43s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eda78fa0048844ae996d42293a2af1e0 : SUCCESS in 2m 49s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff15f3da41a1424998e1e459f8a34084 : SUCCESS in 4m 26s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04131b1ac8c14d49837b710e45e1996a : SUCCESS in 28m 34s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3b18a6e8b5046988cde802950c39052 : SUCCESS in 30m 21s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee938e5147ff470fa745fd44d0b88768 : FAILURE in 31m 55s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5e5802441e84a0a9986f92cc7210ab4 : SUCCESS in 43m 48s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0939f29e3e74a3fb14ba00d1bd43ca6 : FAILURE in 51m 34s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e173ce5060314044a548791ff63e0031 : SUCCESS in 32m 23s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/786967ca0e5f400780312ea641d9852a : SUCCESS in 38m 19s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/89bc4887fe8e42a287c68067c081465b : SUCCESS in 1h 03m 14s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/04dcd67484f0405dafe0ef485ca202e2 : SUCCESS in 55m 39s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/89fdbec2f48f4d40a2201f99e20a9976 : SUCCESS in 55m 49s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ee7273092cd47c7b0f5d10983edf8df : SUCCESS in 31m 31s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5635b5b85a18419fa5a26b59ed4eeb15 : SUCCESS in 1h 24m 14s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f21dab5f5be4f2594082b8b4e431bd4 : SUCCESS in 1h 05m 18s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dce4e79f9c1b4918a6ee628a47d7d997 : SUCCESS in 40m 59s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ae7081804e544ef9cf7dc6307f32094 : SUCCESS in 55m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51ec0224a0204d96a0c06c850c14ebc9 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":3},{"id":"82c883fccd26e022f52bd913b588546c49f3f49f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-10-10 18:14:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"938e014c04a3b4eac2f9ba8722bc1337f8a79135","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2026-03-20 07:29:11.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a9a7c0afc794019e926632c729de1f4a344b2faa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 09:01:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/103a9244ba39487aa0a846f86383fd85\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55cdb700cebb4a6ea12f758912cd4c0a : SUCCESS in 2m 54s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/091c0bdcce5e420c83cff6a29792a418 : SUCCESS in 2m 47s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/fc9a62af49cc4948a818704d27fb6441 : SUCCESS in 4m 31s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e4aed974edc47ed9e7b1445184365d3 : SUCCESS in 4m 46s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c093cfac60d4acead1d65da5a24073c : SUCCESS in 4m 25s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a0c16357cba4105bcc0d68b8e946841 : SUCCESS in 1h 04m 20s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4bcd1f925f84505a84db8fe5de6d742 : SUCCESS in 1h 00m 40s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07eb9fd929f64edeabc87d5573dd798a : SUCCESS in 46m 10s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52a183be691842a4a85fe31dd9036597 : SUCCESS in 45m 09s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36b82c1c86a04215ae51213d4363b1ff : FAILURE in 49m 24s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af9ad09e1f364d498aa46186c1ed5e9a : SUCCESS in 1h 17m 24s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b73bc0b07fe54bed88d597b822b66e14 : SUCCESS in 1h 21m 21s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1258d2ce8be442aaa1297862ce5953b2 : SUCCESS in 51m 37s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1336bfd0f204d8e9b6d280f5db975a9 : SUCCESS in 52m 46s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1057629f6552414599209eb5fe2dc893 : SUCCESS in 59m 28s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d54a064078554443b909c8ceb6ac035c : SUCCESS in 1h 00m 59s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/43a8937967544155afa065a378b97e0a : SUCCESS in 1h 06m 15s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b31511affdea4ae08e54b47499ee7d18 : SUCCESS in 51m 13s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0d6514a76de4f32b0654c1fcfbe4bf2 : SUCCESS in 55m 23s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/27a89f9b3f8b4e9d96d2f8ed0898fd7f : SUCCESS in 54m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29f83be0f1204a5fb1fe247aa5886f11 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":4},{"id":"4735f0fc389c3177f84ab4d74d6a1a25f7ef2291","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-05 15:54:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eb60f5430bfbb9a1b4b272874860a3c547139f83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2026-04-11 08:18:20.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"c970543be33de3749222bcbba56aef3a9255085a","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2026-04-11 08:21:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d20cf9ed8dfd2b44b6eb03c10550bc287d825118","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 09:52:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b150b9b42784fb780ed3f50ea557837\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/000627aae71047e18f470bc2a5b259b7 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3a7c8e684804f5b98b3fa9fd7cfad61 : SUCCESS in 3m 26s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/f4cc195d62a44bd1933cf379a0d7dd0c : SUCCESS in 7m 10s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0f805d046294c9989bddf676c86405a : SUCCESS in 6m 18s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5cbfe4f59f74e6eb2411a0f7ff3e0f9 : SUCCESS in 7m 41s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c7ba904c15e4a7ba8a79bf59e06ebba : SUCCESS in 1h 10m 45s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77d16f51f12e4ccfac7bbf208cb971bf : SUCCESS in 1h 04m 15s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ff7ea03b944467a9844677890484425 : SUCCESS in 53m 34s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46c32ffbe80d419ebddfae7f27cb744a : SUCCESS in 48m 13s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6188442f6b8a432ba8b0fd9ebc9175e0 : FAILURE in 38m 30s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/738df466bf0a45ce977a8d4366a61475 : SUCCESS in 1h 11m 08s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e8923fea320433ab36bc216d03e016f : SUCCESS in 1h 22m 26s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/00d7b6efcbbc4a3796ab52c6df2daee6 : SUCCESS in 1h 01m 06s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e7beb9973074f7389cbfbee0c450de8 : SUCCESS in 54m 58s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6036102d19cd466897a6259d0cd71b6d : SUCCESS in 52m 16s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a795ee720fc43be8f6ed8a4c11b35fa : SUCCESS in 1h 11m 23s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e1c257c2e304e288b355f8ec1371a3d : SUCCESS in 1h 25m 24s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ed8fd2fd9b54e699fbd63b605172658 : SUCCESS in 1h 00m 25s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c9b535df06f4d80a86afb6e6645e8d1 : SUCCESS in 1h 03m 24s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/89c32d1d0e8b456c9eba9fe8b07f837f : SUCCESS in 44m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bbfddbc6e7748beb2dea113b78568fc : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":5},{"id":"20cecafcbba2024506da096ff71a53af5648bced","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2026-04-13 21:07:33.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"dce398b2579be799fa7067ae525b7c783b807b15","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-25 15:45:32.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ccf926f5274f68ccacfa06ad8ec993e63335881f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-25 15:46:37.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"200113be4ed0ae9a42fad686ab17ebcc85102a4d","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2026-05-11 21:33:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"137218e31057b7fb322a6e6e812d18015291eb60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 22:55:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e99165eb8ae47d28589e0a0c68a06bf\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ea7634f16174cdab47af4d60f7f595b : SUCCESS in 1m 51s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96d376da4b2a443e957c80ae8626aa46 : SUCCESS in 3m 14s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/c1f92fd4b7a14aa786af36cb152f6883 : SUCCESS in 5m 12s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adfe1c92630b4812bc80791fc2244803 : SUCCESS in 5m 04s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7e0bfb042fc43a4b1361f2fe7192590 : SUCCESS in 6m 14s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/534bdcb575e1437c80a2deb7306f6dcd : SUCCESS in 29m 17s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1740056ecdaf467b9bc2f530aa9044d3 : SUCCESS in 1h 08m 06s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d415e0a52814ff798a7c287e736b1a3 : SUCCESS in 25m 04s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c180f1b22ad40b59fdf1e8dcff4eaad : SUCCESS in 46m 09s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88fc6c0a004b41089b05459bf00c9b30 : FAILURE in 55m 27s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bb19b9a8cdc402993a6285116efeb5d : SUCCESS in 1h 15m 43s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/384e9ef9d86746c4a6ffbf1b7006c5d8 : SUCCESS in 1h 15m 55s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/61854682d3504e298b4d2334741139ee : SUCCESS in 1h 13m 14s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9188b86e18a41a4914fe87b707dd20a : SUCCESS in 30m 19s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f25ca679a25f4603b01d875355f2346c : SUCCESS in 35m 09s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/56897265cbc04a7b804eb736b8bc2870 : SUCCESS in 1h 02m 19s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2dba9399f75146988e61dd9a28e0633d : SUCCESS in 40m 38s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ec7384d300348ae890f662517d7af62 : SUCCESS in 1h 01m 14s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d43e20b7de84ad7ac934af7f7508875 : SUCCESS in 55m 49s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c357e88613fd4d5381aedff7e3f6279b : SUCCESS in 1h 07m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5aa3f05bc2943ef9df40adaccd5d277 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":5},{"id":"1c634179e866f3ad5dd933aafc03140c4531ac1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 22:56:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2d8f43fa57ea5541af01bf98b2559e06eb49234a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 00:23:57.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6a3b69a391944049d4e1e1781642acc\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/785ed0e85d004018aa262ab18e3f7aca : SUCCESS in 3m 46s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/886a600ed6c24391a78922eeb9645adb : SUCCESS in 3m 51s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/8037986d3a1b4dd198a3648af517fe98 : SUCCESS in 4m 04s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15bce1fa3fd945ac977bff32ded20bc6 : SUCCESS in 6m 01s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f9b74b948634cb190dced30acdae6dd : SUCCESS in 32m 26s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/56503208f3814b6ab6cd17f07b492413 : SUCCESS in 1h 22m 52s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc854a0e74c34c3cbdbc7d06c2aefc26 : SUCCESS in 1h 04m 34s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7112b0425d2e41e2b5db35af77983abc : SUCCESS in 27m 48s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/32b9b49ee00140ae893c9f0e1db3e10a : SUCCESS in 28m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c36d55d29ae4933ade6b7078c349212 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":5},{"id":"ec994abd5b3494088705e9d2f6bacd57ab0587bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 00:23:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"1de289da607d0fcb9f0f2d938c7c4b6084d0d9af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 00:24:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de805cb6e9774736b600c6f40b597fed\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e481578d5f6847d084fe943fe202410d : SUCCESS in 48s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"114e4d76d784b926908e3c321844dd9db261ea55","revisions":{"286dca9145caa7dbb730b4d2dc02897a4150613e":{"kind":"REWORK","_number":1,"created":"2025-10-06 14:03:45.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/86/963186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/86/963186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/1"}}},"commit":{"parents":[{"commit":"25cd83b86f5f42faa417f6915c52026b5b75d72f","subject":"Merge \"Validate ensure parameter for package resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/25cd83b86f5f42faa417f6915c52026b5b75d72f"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2025-10-06 14:02:05.000000000","tz":120},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2025-10-06 14:03:35.000000000","tz":120},"subject":"add posibility to specify db connection parameters","message":"add posibility to specify db connection parameters\n\nChange-Id: I7359f3826c61b058a5f0261e2229b07c31f32e88\nSigned-off-by: Benedikt Trefzer \u003cbenedikt.trefzer@cirrax.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/286dca9145caa7dbb730b4d2dc02897a4150613e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/286dca9145caa7dbb730b4d2dc02897a4150613e"}]},"branch":"refs/heads/master"},"a1a1ef4ca4bca0e811dcc0e92df5da4a6bb8d296":{"kind":"REWORK","_number":2,"created":"2025-10-06 14:35:15.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/86/963186/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/86/963186/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/2"}}},"commit":{"parents":[{"commit":"25cd83b86f5f42faa417f6915c52026b5b75d72f","subject":"Merge \"Validate ensure parameter for package resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/25cd83b86f5f42faa417f6915c52026b5b75d72f"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2025-10-06 14:02:05.000000000","tz":120},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2025-10-06 14:34:22.000000000","tz":120},"subject":"add posibility to specify db connection parameters","message":"add posibility to specify db connection parameters\n\nChange-Id: I7359f3826c61b058a5f0261e2229b07c31f32e88\nSigned-off-by: Benedikt Trefzer \u003cbenedikt.trefzer@cirrax.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/a1a1ef4ca4bca0e811dcc0e92df5da4a6bb8d296"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/a1a1ef4ca4bca0e811dcc0e92df5da4a6bb8d296"}]},"branch":"refs/heads/master"},"ce75d841d32ee3764233f5f5679f951d4572d269":{"kind":"REWORK","_number":3,"created":"2025-10-06 15:17:34.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/86/963186/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/86/963186/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/3"}}},"commit":{"parents":[{"commit":"25cd83b86f5f42faa417f6915c52026b5b75d72f","subject":"Merge \"Validate ensure parameter for package resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/25cd83b86f5f42faa417f6915c52026b5b75d72f"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2025-10-06 14:02:05.000000000","tz":120},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2025-10-06 15:17:19.000000000","tz":120},"subject":"add posibility to specify db connection parameters","message":"add posibility to specify db connection parameters\n\nChange-Id: I7359f3826c61b058a5f0261e2229b07c31f32e88\nSigned-off-by: Benedikt Trefzer \u003cbenedikt.trefzer@cirrax.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/ce75d841d32ee3764233f5f5679f951d4572d269"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/ce75d841d32ee3764233f5f5679f951d4572d269"}]},"branch":"refs/heads/master"},"751bb41c4010f0e330754ea6d60025c9a68f4e78":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-20 07:29:11.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/86/963186/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/86/963186/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/4"}}},"commit":{"parents":[{"commit":"9cd4028fb32b1d2474eee33cf7957bdd6abaf053","subject":"cache: Deprecate support for legacy backend specific options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9cd4028fb32b1d2474eee33cf7957bdd6abaf053"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2025-10-06 14:02:05.000000000","tz":120},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2026-03-20 07:29:02.000000000","tz":60},"subject":"add posibility to specify db connection parameters","message":"add posibility to specify db connection parameters\n\nChange-Id: I7359f3826c61b058a5f0261e2229b07c31f32e88\nSigned-off-by: Benedikt Trefzer \u003cbenedikt.trefzer@cirrax.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/751bb41c4010f0e330754ea6d60025c9a68f4e78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/751bb41c4010f0e330754ea6d60025c9a68f4e78"}]},"branch":"refs/heads/master"},"114e4d76d784b926908e3c321844dd9db261ea55":{"kind":"REWORK","_number":5,"created":"2026-04-11 08:18:20.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/86/963186/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/86/963186/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/86/963186/5"}}},"commit":{"parents":[{"commit":"9cd4028fb32b1d2474eee33cf7957bdd6abaf053","subject":"cache: Deprecate support for legacy backend specific options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9cd4028fb32b1d2474eee33cf7957bdd6abaf053"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2025-10-06 14:02:05.000000000","tz":120},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2026-04-11 08:18:31.000000000","tz":120},"subject":"add posibility to specify db connection parameters","message":"add posibility to specify db connection parameters\n\nChange-Id: I7359f3826c61b058a5f0261e2229b07c31f32e88\nSigned-off-by: Benedikt Trefzer \u003cbenedikt.trefzer@cirrax.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/114e4d76d784b926908e3c321844dd9db261ea55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/114e4d76d784b926908e3c321844dd9db261ea55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
