)]}'
{"id":"openstack%2Fpuppet-oslo~967710","triplet_id":"openstack%2Fpuppet-oslo~master~I3df4f4cabed4c0031ed0e7bf22d3f10d261b70d6","project":"openstack/puppet-oslo","branch":"master","topic":"hashclient-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-11-28 09:58:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3df4f4cabed4c0031ed0e7bf22d3f10d261b70d6","subject":"cache: Follow renaming of hashclient parameters","status":"MERGED","created":"2025-11-19 16:06:17.000000000","updated":"2025-11-28 09:59:05.000000000","submitted":"2025-11-28 09:58:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967710-hashclient-opts","meta_rev_id":"de4d220063b581ff1ffd960fcc44d8be90fee565","_number":967710,"virtual_id_number":967710,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-28 09:58:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-11-26 15:04:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-11-28 08:37:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-11-28 08:37:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-19 17:24:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-26 15:04:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-11-28 08:37:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"f50d1defea90d5a73c1c0dd92d524253b786f37e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-19 16:06:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c24e871199cb21df2629a840a409c5eebae7c1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 17:24:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a43b35a85a384cdd9986b62149bf7a13\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa6a43a3347d4c6494051d5814544d1f : FAILURE in 3m 42s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fac94edf5b724d5ea80bda849e5bb579 : SUCCESS in 2m 43s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/167c3fe240a54ad1aa76a3d86e929d4a : FAILURE in 6m 01s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f624c3f7a364f38a89c6ca617b09965 : FAILURE in 3m 36s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a43f9c96fdfc46d1ae0a28609916eb0f : FAILURE in 4m 26s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/418f182227ad485ea4b549e621f9f3ff : SUCCESS in 45m 50s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7378222141754e2499a18a05b2523838 : SUCCESS in 59m 39s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4658406060e542299de29f3a337d90e0 : SUCCESS in 51m 19s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c628411a0f442efb68149ddfa27108b : SUCCESS in 46m 09s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/973fe228858f42149a7881285f676524 : FAILURE in 34m 09s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f6bca39903a4bde93a63e29b9962849 : SUCCESS in 1h 05m 48s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a699e0b6208648e09c298137d8373400 : SUCCESS in 52m 40s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ef7faf8b4934e84ace7d6c041caafe4 : SUCCESS in 55m 45s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7597e26fe7264233bad203ab05f1fbdf : SUCCESS in 43m 10s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2217862f8db44fa4bcc1977a606af4f1 : SUCCESS in 54m 42s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3904db118814bd8bcd95a673a4cad81 : SUCCESS in 1h 03m 32s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5bc32e68c4214ef9bb9144751d80533c : SUCCESS in 1h 13m 02s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fddb4dedb0a945769933fb92c19a8761 : SUCCESS in 57m 25s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/29508601e63e4cc5b824100afbd76aa1 : SUCCESS in 35m 18s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b668594073314ca4a8eded4ad038ff44 : SUCCESS in 54m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b9bc35c2c6a45f7a60a433d0798bb90 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"10dc1e723ae5d4b4fcc7448f8cdcd9591129f219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-19 17:48:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e8edd4744b872164f9c22ce2da3b06bf7c1dfdca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-19 17:49:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1a6c1d466963069632dcf44b3189333d939d41c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 19:17:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/109e2041bed94025abe0be122550bc42\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb0650961a984feeb22b6ef9080f005f : FAILURE in 2m 18s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd917ac3a8164a95a859c2dcf7f2d81c : SUCCESS in 2m 14s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/9ad7dd5420dd43928570738fff504aec : SUCCESS in 3m 48s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6e7a0911e054cb8b72e98249f88d5ca : SUCCESS in 4m 12s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14d819a8a90b4163b1332ffc8dad2895 : SUCCESS in 3m 14s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99d032ce93da461ea66d55e2a61da6ac : SUCCESS in 54m 28s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/129ae3a6519b4d5eb5279f4dda82da26 : SUCCESS in 1h 15m 41s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90be5828033e44c59fbae069e89ffa62 : SUCCESS in 46m 43s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67bc802dd26c43999e7716c5a8fdae0b : SUCCESS in 44m 49s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d80a3475fa1441688fd20b98115d779 : FAILURE in 58m 06s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/931e2b3bbf594bf7b4cc05aee699b9fa : SUCCESS in 51m 22s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfbfebd06cd84a559ca446783dca6a80 : SUCCESS in 1h 00m 45s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/85f513a20f054aba880eef4abd09abdd : SUCCESS in 45m 50s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a61a27099f7d4ad58c5b449acc060b29 : SUCCESS in 30m 17s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1496357c66e447f08fdafc29c753f41b : SUCCESS in 32m 33s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/354bb93909084e6d94769c72babaf1fb : SUCCESS in 1h 09m 26s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ef790a088684f72a612e3ee70f70111 : SUCCESS in 1h 23m 33s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fd75e606168434288d6026d8272ec97 : SUCCESS in 29m 38s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc67bd330d924445b772a0f975db2142 : SUCCESS in 29m 55s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/74a30cc2706c497d93d318780cda4fc3 : SUCCESS in 33m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d200cfcab728457f835e69c7dc7ef001 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":3},{"id":"2f591bffc19c60728a2d20e291a2bd4290e5b557","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-20 02:26:39.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"917ce1d2102adadc7dc16caebc4b6617340c6fe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 03:55:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eec631b7a49a427d9fd8e9cce2508cfb\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e55183ae456c47eeb96fd2ed5467c559 : FAILURE in 4m 26s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/256a730f7fad4162b20ebe14779f6ea9 : SUCCESS in 3m 16s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/e16c5757fd6e46f9ac19e67750097f1d : SUCCESS in 5m 28s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2ab92cd6a824cfcab2a548ceedbe4e1 : SUCCESS in 4m 17s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a56c90d28d34dfebfecf631ec2d62d3 : SUCCESS in 4m 43s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/079f3ed909d0491abdc3fc808e7b1a1a : SUCCESS in 56m 46s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f3cb2d92f744f9abce06411a542655a : SUCCESS in 59m 47s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94a9bcc5000b446abc2356a2d7122961 : FAILURE in 40m 01s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b46d2d54ec64c7e88839a63da872b0b : SUCCESS in 43m 56s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96feb416d59940bcb34ff15b608a7e72 : FAILURE in 54m 06s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/db5463f95f5146b0ba2133ef8181fdee : SUCCESS in 1h 20m 42s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c42ab96b1c8424ab4255c37add8dd63 : SUCCESS in 1h 19m 58s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/13f00009dfed46d9851d55828699e318 : SUCCESS in 30m 02s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b14d4efbc60846a7951f2a18299377f9 : SUCCESS in 32m 28s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf9f8dc5bb5e4e7a8fcf1145574ca975 : SUCCESS in 51m 53s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b680f56470894cfc9895fa98309e0cf9 : SUCCESS in 1h 06m 34s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/773656e67cd94199867a3f034d35313a : SUCCESS in 1h 21m 00s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6f50b41aa1148989e48d645916afe3e : SUCCESS in 50m 22s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6b9dde7b4a146c999106930cfc2cc5f : SUCCESS in 51m 37s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/390a9ed6b1c14da2bde11da40bdc278e : SUCCESS in 34m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8d21ec031e54263a20c3ddeecb3213d : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":4},{"id":"1a2ccae55b657417970764228254ef0f057f8c32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-20 11:16:38.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"da4fe4cab28c64c7c91c6585935ac4f6fc9a14cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 12:50:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f121a8528114435a3f6e6f6bfc50d50\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26ba0666d57d440eb85c07901a4ebf70 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd67f36adc134652bd81f3a9f41bd30b : SUCCESS in 2m 16s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/e68cc12a5e5d48d398a03ebf7457fa73 : SUCCESS in 5m 06s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cbc2788cac94dd0b264069ac4755f9d : SUCCESS in 2m 48s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/006ddd76b374489bbae41bab386a7a24 : SUCCESS in 2m 55s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d92e2400624412d9b92abe4053643c2 : SUCCESS in 57m 12s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/904dfeb857994ea7be2d08673546c1c2 : SUCCESS in 34m 02s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2287b07fc414843a8e1478a1a71c47d : SUCCESS in 59m 38s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c32c9186f78b4516b2b9bf62d9169f34 : SUCCESS in 51m 53s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7a79bcecf3c4cffa2047fec289e6a86 : FAILURE in 54m 43s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aac3fc016a244954b5b090eaaee6f147 : FAILURE in 1h 24m 56s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/34ee526ba423458cb598bd6a80ab47a1 : SUCCESS in 44m 26s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3d156ffc88d48b8b4fdd4dcb6f02b2d : SUCCESS in 32m 54s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/afa8a30af441417586c538e586309b14 : SUCCESS in 56m 39s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae565eeaf361406fb004ffa3a024dc92 : SUCCESS in 58m 20s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/17590c8e04e742aa90af205c3b236e85 : FAILURE in 1h 27m 50s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/19b7e10d54444ddcbf78f2a0e183256c : SUCCESS in 1h 18m 19s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a28ee6e3d1a4ed38aa42dd939966655 : SUCCESS in 1h 09m 41s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/405b65bd212a47939b154790e8b370b8 : SUCCESS in 52m 19s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/17ad01cf6a99497d946efc4bac29d7bc : SUCCESS in 27m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a6f50b567b646d0925bfb5b19818592 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":5},{"id":"4e5086fa18b45e3d651b629db867bf02dc7acba1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-20 14:37:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4354451638cf6e07aa6427f2241758de794b4f99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 16:13:18.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fbecad5e4d147a8b898353d727dc6c1\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ff9b42778a84d1fb27db156f4755d40 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3912a9bc2c740fe99a77d80a8ae4989 : SUCCESS in 1m 49s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/2ac3871cb8f249d3ad80cc6142bf11eb : SUCCESS in 5m 04s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9af506378c48400da21599ced9a1e7b4 : SUCCESS in 4m 39s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddaf459c67354c64ae2c1641ce1011e5 : SUCCESS in 3m 11s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d1b36a8ebe54d3f97bc9f8235f301cc : SUCCESS in 52m 52s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/234423f98eea487caffb3d522a36fbee : SUCCESS in 1h 03m 00s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2711d2e0de4d4da5bbd11de8b9705d5c : SUCCESS in 38m 01s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ee9ff4908164f5e8c3c7d10c25fc88d : SUCCESS in 49m 14s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5f18e7386e44bc09e9f3537e0f8947f : FAILURE in 53m 41s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec9b5ac3ce2540df82ffdcd7da4912b1 : FAILURE in 1h 27m 58s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f11fe05cfbdf47cb82fa0c59d38fcb8f : SUCCESS in 1h 17m 34s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/89f624f91dca4087a1bd73d5be61e4e4 : SUCCESS in 26m 44s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c917a5acbb564358a904dbbd187cd6ed : SUCCESS in 53m 21s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ebe3f835f9024a75bf2f47d0f5b96d7c : SUCCESS in 44m 36s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/347264c7c29b43a7970c0ac13fdb032f : FAILURE in 1h 28m 51s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/336896123d6e4b819273192a19107e4d : SUCCESS in 1h 30m 06s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/317bdd652d8b4860a508b73bf7a48005 : SUCCESS in 1h 01m 00s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fca83e8ba094d50adffbc49100f296d : SUCCESS in 44m 01s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/97d90b33738f4c5db3317c9cd3f5f849 : SUCCESS in 27m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c7d6df706294d3ba49486f1ac17a52f : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":5},{"id":"4267f65a436d6c2151f6c9417bc77212120ec16d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-22 15:40:36.000000000","message":"Topic set to hashclient-opts","accounts_in_message":[],"_revision_number":5},{"id":"360721b05a7c1795b2155b4d2d2bcfe7c5f17039","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-24 08:52:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"81e9d76e2e3f86663c7a8d8d622d1e92072ec4b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 10:26:10.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/226b4d6a4ede4b16a4fbb84d77a5e256\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21b01d94878a4dfdb9bdc44211fa8e2b : SUCCESS in 2m 09s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15d580ba43384aacbda4551bbc06da08 : SUCCESS in 3m 24s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/713fcd9483ab48a5809297b58421b2cd : SUCCESS in 3m 49s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12ecc5e2f83848b5a9180c904cfee8bc : SUCCESS in 4m 45s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cbb816bb89549c5ac51f2abd626d26c : SUCCESS in 4m 47s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5050a9f1243448e5a57ab05f79e36ef9 : SUCCESS in 55m 10s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c30728aaad4645ebbd09f90a4d1ca7b8 : SUCCESS in 1h 05m 59s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/484a6c95d6c544aeaaf82e75eb8b04e0 : SUCCESS in 23m 33s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/460e130819954f38ad52e50a921d7281 : SUCCESS in 23m 45s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2badb7ee3a34433b924e2d42316db304 : FAILURE in 48m 04s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/edd536a796734bbaa71b1f48a13711b5 : FAILURE in 1h 27m 16s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f349b628bb614a3c855021348c13e9e3 : SUCCESS in 1h 26m 43s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ccb56d266a704497a2b73db289c323a3 : SUCCESS in 59m 20s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a99772ba24df40bd9412d8c744e47d39 : SUCCESS in 58m 48s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cdd42ac7be644f87b072ed361bf27223 : SUCCESS in 29m 54s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4fda1bf7c224ff5b7cf31b7fa0373f1 : FAILURE in 1h 27m 41s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e860838165f54b6ab415f6840dc2ec07 : SUCCESS in 1h 21m 36s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d10fd3b15cd140f7907a7f652b9b2f29 : SUCCESS in 1h 01m 38s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6bc810f40c94b2db90f81eaa8bc11cc : SUCCESS in 54m 01s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc7426a2d9c04e3c82c48ba1e4dcaeff : SUCCESS in 53m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f485037c1a3b4b0c9f5b6447e8e807e1 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":5},{"id":"211344751dc73dda9a501a1df890bc1d8c2599f5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-25 03:14:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"abf6009b085c745fa7e91d2432f20e150927e580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 04:45:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/431e3c0c0cf8489f895be36698343396\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/206af57e8742444c926a45c4eb35cf05 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be63733aa80549ef8c210ff6a490b42c : SUCCESS in 2m 14s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/6609a99753ca4ce4ab2d3353c84ee711 : SUCCESS in 3m 56s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecd88d6fdb6c40ec9507dba11801861f : SUCCESS in 2m 49s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac4cf1d99808488180c5e76f9ec7669c : SUCCESS in 4m 22s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c45f1930c90400dbec25b59ec656c29 : SUCCESS in 1h 03m 15s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c838cc10ee484df68c61143174bc4eda : SUCCESS in 34m 55s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2b15162849343f4b4383f4c80d04e60 : SUCCESS in 23m 16s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d135ac9ee0264c69b7bb7b463b51eb75 : SUCCESS in 45m 01s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c4a18ef6c054fe68d7b175151a2cf7b : FAILURE in 30m 43s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/170850a8ec8d471fa98d612c95b0532e : SUCCESS in 33m 53s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e031540574e47bebce7e6f07511afbe : SUCCESS in 1h 19m 38s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/44c18d94decc4c15993e9f071845d096 : SUCCESS in 29m 20s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c86cb16dac514b1192522eb0728b63eb : SUCCESS in 54m 44s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/89932877de6c461697244d2e650bd92e : SUCCESS in 1h 03m 00s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c424d864a80f4fd58c2ee6f92617fbe3 : SUCCESS in 1h 10m 31s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a2fa5cc3be24316a5d7f872b3c02f2b : SUCCESS in 1h 25m 11s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee6781003d0a4fb39d81f9ccf4fccd26 : SUCCESS in 1h 01m 44s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/47f3da843f0941c8937275d1d4d0a03e : SUCCESS in 48m 34s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7cb76fbfdd8f45e18f7e39d796301bb3 : SUCCESS in 30m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fc05a5a2be14d3e85f87d935a707074 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":5},{"id":"1cf55a83cde781dd28eb471fb63d87d1e72aa109","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-11-26 15:04:24.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"2c4a7ef9ca6c430c56040931b5b6c6c86292e681","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-11-28 08:37:16.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"54c91dc962bbe6da8cfcb93e6afd826edf77ec90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 08:37:59.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"37345118be707b9b39f4933f96e734d515793567","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 09:58:21.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cac312a6a3f485c95cb07ca24ea142a\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81e691c0a3c74eb1b6ca89ea6650a9bc : SUCCESS in 2m 54s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c890af7c62b0483883f0de363dadec4a : SUCCESS in 3m 34s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/e1ae57268807484c908fc23815d59ad5 : SUCCESS in 3m 42s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87bb24f934914771a4ffc7eaf31a3a60 : SUCCESS in 4m 31s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e365d33ff4af46c1b601c0ea3efc5d0b : SUCCESS in 57m 07s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/267f7b19b23d4c86a80d87d33a17d178 : SUCCESS in 1h 16m 04s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/84f05b12da9740adb828eba77a1a0ec3 : SUCCESS in 57m 19s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/54cb3554431c4443b46bb811131060e9 : SUCCESS in 28m 36s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6593241f0de947248497fc50e2554abc : SUCCESS in 55m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b96eccfdb7cd495dabecbd77d1a2aa12 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":5},{"id":"57c6875b7701107eae40c93e5df09c13515c3079","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 09:58:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"de4d220063b581ff1ffd960fcc44d8be90fee565","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 09:59:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc0306121f8d4077a4b36eea870623a8\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/714a3e84f9024b7fa0765eaf43d3ae9a : SUCCESS in 35s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5233475e5a4c6be7bbc01b8742eb36940f584bbe","revisions":{"49cb677e7abc1459b4feec804bd779a60eb6dc9f":{"kind":"REWORK","_number":1,"created":"2025-11-19 16:06:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/967710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/10/967710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/1"}}},"commit":{"parents":[{"commit":"c671097562200bfb041373919a568c0d4cdecb2e","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c671097562200bfb041373919a568c0d4cdecb2e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-19 16:05:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-19 16:06:04.000000000","tz":540},"subject":"cache: Adopt to renaming of hashclient parameters","message":"cache: Adopt to renaming of hashclient parameters\n\nDepends-on: https://review.opendev.org/967580\nChange-Id: I3df4f4cabed4c0031ed0e7bf22d3f10d261b70d6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/49cb677e7abc1459b4feec804bd779a60eb6dc9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/49cb677e7abc1459b4feec804bd779a60eb6dc9f"}]},"branch":"refs/heads/master"},"c2af9134401afe6e8dc34908210ac3204164afd2":{"kind":"REWORK","_number":2,"created":"2025-11-19 17:48:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/967710/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/10/967710/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/2"}}},"commit":{"parents":[{"commit":"c671097562200bfb041373919a568c0d4cdecb2e","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c671097562200bfb041373919a568c0d4cdecb2e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-19 16:05:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-19 17:48:08.000000000","tz":540},"subject":"cache: Adopt to renaming of hashclient parameters","message":"cache: Adopt to renaming of hashclient parameters\n\nDepends-on: https://review.opendev.org/967580\nChange-Id: I3df4f4cabed4c0031ed0e7bf22d3f10d261b70d6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c2af9134401afe6e8dc34908210ac3204164afd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c2af9134401afe6e8dc34908210ac3204164afd2"}]},"branch":"refs/heads/master"},"8beb08033164add4d6dccafdeea2d51554d21827":{"kind":"REWORK","_number":3,"created":"2025-11-19 17:49:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/967710/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/10/967710/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/3"}}},"commit":{"parents":[{"commit":"c671097562200bfb041373919a568c0d4cdecb2e","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c671097562200bfb041373919a568c0d4cdecb2e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-19 16:05:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-19 17:49:14.000000000","tz":540},"subject":"cache: Follow renaming of hashclient parameters","message":"cache: Follow renaming of hashclient parameters\n\nDepends-on: https://review.opendev.org/967580\nChange-Id: I3df4f4cabed4c0031ed0e7bf22d3f10d261b70d6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/8beb08033164add4d6dccafdeea2d51554d21827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/8beb08033164add4d6dccafdeea2d51554d21827"}]},"branch":"refs/heads/master"},"495cc4409e97076d2985fda604863216596f49b6":{"kind":"REWORK","_number":4,"created":"2025-11-20 02:26:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/967710/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/10/967710/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/4"}}},"commit":{"parents":[{"commit":"c671097562200bfb041373919a568c0d4cdecb2e","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c671097562200bfb041373919a568c0d4cdecb2e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-19 16:05:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-20 02:26:39.000000000","tz":0},"subject":"cache: Follow renaming of hashclient parameters","message":"cache: Follow renaming of hashclient parameters\n\nDepends-on: https://review.opendev.org/967580\nChange-Id: I3df4f4cabed4c0031ed0e7bf22d3f10d261b70d6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/495cc4409e97076d2985fda604863216596f49b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/495cc4409e97076d2985fda604863216596f49b6"}]},"branch":"refs/heads/master"},"5233475e5a4c6be7bbc01b8742eb36940f584bbe":{"kind":"REWORK","_number":5,"created":"2025-11-20 11:16:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/967710/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/10/967710/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/10/967710/5"}}},"commit":{"parents":[{"commit":"c671097562200bfb041373919a568c0d4cdecb2e","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c671097562200bfb041373919a568c0d4cdecb2e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-19 16:05:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-20 11:16:38.000000000","tz":0},"subject":"cache: Follow renaming of hashclient parameters","message":"cache: Follow renaming of hashclient parameters\n\nDepends-on: https://review.opendev.org/967580\nChange-Id: I3df4f4cabed4c0031ed0e7bf22d3f10d261b70d6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/5233475e5a4c6be7bbc01b8742eb36940f584bbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/5233475e5a4c6be7bbc01b8742eb36940f584bbe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
