)]}'
{"id":"openstack%2Fpuppet-oslo~993266","triplet_id":"openstack%2Fpuppet-oslo~master~I377cb93b625206ee7b0113a4e09a3d280c5815eb","project":"openstack/puppet-oslo","branch":"master","topic":"remove-postgresql","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-05 02:30:35.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I377cb93b625206ee7b0113a4e09a3d280c5815eb","subject":"Remove PostgreSQL support","status":"MERGED","created":"2026-06-14 15:58:29.000000000","updated":"2026-07-05 04:11:01.000000000","submitted":"2026-07-05 04:10:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993266-remove-postgresql","meta_rev_id":"52f9787137961486262dc3f1c1a22d499115d793","_number":993266,"virtual_id_number":993266,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-05 04:10:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-07-05 02:30:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-06-30 06:13:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-07-05 02:30:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-14 17:24:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-30 06:13:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"f215363699096b0e9f3cc9eedae4e628046a9023","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-14 15:58:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2605a81256029eea1b7bd8c2358adce79c5901ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-14 15:59:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1fcbc247fd37ed507cc9133c513d113238a35ccd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-14 17:24:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a6a85ccca024e388ef23db06a6c064d\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f2ce3a60d3a4d989aed81feaf73b545 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3695b653a2664245bdc8f7707b66a2c8 : SUCCESS in 2m 43s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/574c1007da6b4d16ad897805ab95abe2 : FAILURE in 4m 15s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec9191507f2548638b15ed360f018ae3 : FAILURE in 2m 59s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff92ff01af72462499c70c0d0c5db9db : FAILURE in 7m 40s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f35d69d14a544565bd7dee6bd4964eed : SUCCESS in 1h 00m 45s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8efe0c8a09d04647b931c1b04501ca08 : FAILURE in 1h 10m 21s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5e3b31778064f23b512b2909e11b4a3 : FAILURE in 1h 03m 40s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a69d40ab50944dc93cfb0e267a22a18 : SUCCESS in 24m 17s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42f570e33cff4ee7919b7aa7e3bfb012 : FAILURE in 51m 56s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5cfbec99a3c4400ea02e2ed93ca7d991 : SUCCESS in 45m 48s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c6c70ba7a7c45a191223b3f751de824 : SUCCESS in 56m 24s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7fe2a31e290840ec8c0903b16c8d2829 : SUCCESS in 59m 02s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3270c2efab0140e8b278bf84c2344068 : SUCCESS in 56m 47s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ceec20c159c407b9003989f718b06b0 : SUCCESS in 28m 55s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a1434f7c85d4c81bd54a2f446f0d0b7 : SUCCESS in 32m 59s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/acdc75a451ca4a289bf2eb06958855d4 : SUCCESS in 1h 20m 17s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2a41603f8f047c8b6d2d1598b5421f2 : SUCCESS in 45m 09s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/309db3d6ddf548b9b30c1c2f28f85d90 : SUCCESS in 56m 56s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/29a919d7dd7c4161bab6ecf448de687b : SUCCESS in 55m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1376c2b7f9f4970b901be12938eff1e : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":2},{"id":"bb65fa246f45c50791ffbd77f257144e660680ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-15 07:38:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"01d084d8f0ffa2e2e096a8bc81add6382c4cd801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 09:04:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a582b5f8a955471282fa51074b63f1e2\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5164f20879614f6aa06caac5140787ea : SUCCESS in 3m 45s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f554b38f8c1345c0a04bd5855e59d360 : SUCCESS in 1m 51s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/8ed7b8d2930f4b6291feb52219d33b0e : SUCCESS in 6m 05s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87a6c7057f264702b1f2952592807178 : SUCCESS in 5m 59s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0419bfc30e0940c3a2e089b50a61ac0c : SUCCESS in 4m 32s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c84f4c400d054d208cbf6cb98ec9225f : SUCCESS in 1h 02m 09s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8424217b1abd4ce0b97167d3234a6f0b : FAILURE in 1h 12m 35s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e9e4b6255d144e28cbc5f260b8172fc : FAILURE in 1h 19m 13s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38a5399f8e8e483bb7c427c687a96762 : SUCCESS in 47m 59s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1694a7e808141bd950110ee12726894 : FAILURE in 1h 01m 36s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/17c475d5e4eb4933ab56f1152291d70d : SUCCESS in 1h 04m 55s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e68f58a13acc43aca276422a5b065fe0 : SUCCESS in 1h 17m 03s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6acd742f60bc4115a19acfd7395b2f83 : SUCCESS in 1h 00m 22s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/817dddb193d247b5beabc56a253ad8b4 : SUCCESS in 1h 02m 09s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8afa763cd64d4d659748a24d26dda8d9 : SUCCESS in 52m 01s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/38329f4f71bb484aaa1b74c7b0e706a7 : SUCCESS in 1h 02m 29s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e9b272500b648bcb975de14a790c312 : SUCCESS in 1h 12m 46s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4da51379327e40c5ae76449f14af7903 : SUCCESS in 58m 53s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce070cafdc35415b9ee9693eefe6f092 : SUCCESS in 57m 18s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7d192442b664aea846c2a6586eeff7e : SUCCESS in 43m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf5f20734442449080a34caba5029559 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"0a0b3546943e36b636dc922a5e68808ab8310b0c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 06:13:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c8a41c9436712b12080e362acdca45bf9e4de65a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-04 13:33:53.000000000","message":"Topic set to remove-postgresql","accounts_in_message":[],"_revision_number":3},{"id":"074c1ca4ccd4b465fe427c6240d1abab457aadf1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-05 02:30:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"61cc23a58ea317e958468c4350e5a98d2c1158cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 02:31:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"80d5cc583bd977d515ddc9a005869156ae44809a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 04:10:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce6cb42dd3d944fe983995a1b58967a5\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2804a56a18db4e0ea9329be0eac7b8e3 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df954e7f70af4c38af22a4304daff6c2 : SUCCESS in 2m 56s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/f3aabe1662e4430390b92afec5d4b899 : SUCCESS in 18m 06s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c41498f2bf2441d4b51e037412a09eca : SUCCESS in 6m 55s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/273ef00a03b6408c8caf7b18dfbb266d : SUCCESS in 1h 03m 32s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ccda6ebb0474aaa95979e2c5911d6a4 : SUCCESS in 1h 30m 40s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/730ed04302ad43d3839c1e8dfa59b4ad : SUCCESS in 31m 17s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/67297595eb9543d7af6b49c6d68bc564 : SUCCESS in 1h 02m 44s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3fc4adda6896408d890bc0ac0897ad58 : SUCCESS in 42m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29d9501350574cefae70f285554b27ec : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":3},{"id":"201340eab5e0bfd3f835d7940f4ecff88dd52b39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 04:10:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"52f9787137961486262dc3f1c1a22d499115d793","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 04:11:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab09184c867c44e8b52f1487621d9eb5\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8baf534a74ab45b4b931cecaa912e6e6 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bb284ccbde0c433c2cf964cada24311934ffebda","revisions":{"38b4f4e576ec22d9adbd562cc61eee9d593e3352":{"kind":"REWORK","_number":1,"created":"2026-06-14 15:58:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/993266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/66/993266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/993266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/993266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/993266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/66/993266/1"}}},"commit":{"parents":[{"commit":"3e09d81575f6e92283428160b30a38d7b5ff224b","subject":"Merge \"Bump upper version of puppetlabs-stdlib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3e09d81575f6e92283428160b30a38d7b5ff224b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-14 15:58:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-14 15:58:09.000000000","tz":540},"subject":"Remove PostgreSQL support","message":"Remove PostgreSQL support\n\nIt was deprecated during the previous cycle[1], due to neutron which\nalready removed PostgreSQL support.\n\n[1] 0c4812f8815812a0e938089239f5a4964f8e37c4\n\nChange-Id: I377cb93b625206ee7b0113a4e09a3d280c5815eb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/38b4f4e576ec22d9adbd562cc61eee9d593e3352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/38b4f4e576ec22d9adbd562cc61eee9d593e3352"}]},"branch":"refs/heads/master"},"63007690c20ef90306889c3882ce533914573b07":{"kind":"REWORK","_number":2,"created":"2026-06-14 15:59:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/993266/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/66/993266/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/993266/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/993266/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/993266/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/66/993266/2"}}},"commit":{"parents":[{"commit":"3e09d81575f6e92283428160b30a38d7b5ff224b","subject":"Merge \"Bump upper version of puppetlabs-stdlib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3e09d81575f6e92283428160b30a38d7b5ff224b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-14 15:58:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-14 15:59:37.000000000","tz":540},"subject":"Remove PostgreSQL support","message":"Remove PostgreSQL support\n\nIt was deprecated during the previous cycle[1], due to neutron which\nalready removed PostgreSQL support.\n\n[1] 0c4812f8815812a0e938089239f5a4964f8e37c4\n\nChange-Id: I377cb93b625206ee7b0113a4e09a3d280c5815eb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/63007690c20ef90306889c3882ce533914573b07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/63007690c20ef90306889c3882ce533914573b07"}]},"branch":"refs/heads/master"},"bb284ccbde0c433c2cf964cada24311934ffebda":{"kind":"REWORK","_number":3,"created":"2026-06-15 07:38:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/993266/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/66/993266/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/993266/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/993266/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/993266/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/66/993266/3"}}},"commit":{"parents":[{"commit":"3e09d81575f6e92283428160b30a38d7b5ff224b","subject":"Merge \"Bump upper version of puppetlabs-stdlib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3e09d81575f6e92283428160b30a38d7b5ff224b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-14 15:58:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-15 07:38:48.000000000","tz":540},"subject":"Remove PostgreSQL support","message":"Remove PostgreSQL support\n\nIt was deprecated during the previous cycle[1], due to neutron which\nalready removed PostgreSQL support.\n\n[1] 0c4812f8815812a0e938089239f5a4964f8e37c4\n\nChange-Id: I377cb93b625206ee7b0113a4e09a3d280c5815eb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/bb284ccbde0c433c2cf964cada24311934ffebda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/bb284ccbde0c433c2cf964cada24311934ffebda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
