)]}'
{"id":"openstack%2Fpuppet-ovn~654199","triplet_id":"openstack%2Fpuppet-ovn~master~I9d06eeeb32b6d7d81312ecff71a848c58f903eff","project":"openstack/puppet-ovn","branch":"master","topic":"move-to-opendev","hashtags":[],"change_id":"I9d06eeeb32b6d7d81312ecff71a848c58f903eff","subject":"Move to opendev","status":"MERGED","created":"2019-04-21 06:45:42.000000000","updated":"2019-04-22 17:30:13.000000000","submitted":"2019-04-22 17:30:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654199-1555954213667-795c03e7","meta_rev_id":"7ccedd9cd799dcdaaffc044ff57abbee7dd8f9a4","_number":654199,"virtual_id_number":654199,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-04-22 17:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-04-22 17:05:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-04-22 17:05:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 17:05:35.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-04-22 17:30:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a2244c2cf64465b85204262de324a9c1c8a6d17","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 06:45:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e95362492f377c0a06ba9a232d203a76d2012da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 07:17:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/99/654199/1/check/puppet-openstack-lint-ubuntu-bionic/37553e9/ : SUCCESS in 7m 31s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/99/654199/1/check/puppet-openstack-syntax-4-ubuntu-bionic/bcb0705/ : SUCCESS in 7m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/99/654199/1/check/puppet-openstack-syntax-5-ubuntu-bionic/bae20ec/ : SUCCESS in 5m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/99/654199/1/check/puppet-openstack-syntax-6-ubuntu-bionic/b321d92/ : SUCCESS in 5m 45s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/99/654199/1/check/puppet-openstack-unit-4.8-centos-7/ab75f01/ : SUCCESS in 8m 57s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/99/654199/1/check/puppet-openstack-unit-5.5-centos-7/55865b5/ : SUCCESS in 7m 43s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/99/654199/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/a236700/ : SUCCESS in 8m 38s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/99/654199/1/check/puppet-openstack-unit-latest-ubuntu-bionic/cebdb5a/ : FAILURE in 6m 20s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/99/654199/1/check/puppet-openstack-beaker-centos-7/e8a08d0/ : SUCCESS in 10m 58s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/99/654199/1/check/puppet-openstack-beaker-ubuntu-bionic/fe05762/ : SUCCESS in 9m 30s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/99/654199/1/check/build-openstack-releasenotes/9bab659/html/ : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"fc2a4a3bb2aae4bbdfed1ddf05022f872048361d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-22 17:05:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c5614a764a2c3d1772cecc5c333a5a6fc066926","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-22 17:05:35.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f96daf5f54cf932a29f62e44110085f6ae8a12a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:05:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16ac89bcb3720e9d0dcea4517eb66e038a26c799","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:30:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/99/654199/1/gate/puppet-openstack-lint-ubuntu-bionic/b59be2c/ : SUCCESS in 4m 29s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/99/654199/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/05bff62/ : SUCCESS in 4m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/99/654199/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/89bef2a/ : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/99/654199/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/2aaee97/ : SUCCESS in 4m 15s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/99/654199/1/gate/puppet-openstack-unit-4.8-centos-7/add28c0/ : SUCCESS in 9m 24s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/99/654199/1/gate/puppet-openstack-unit-5.5-centos-7/90e04d9/ : SUCCESS in 8m 49s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/99/654199/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/ec75a80/ : SUCCESS in 8m 39s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/99/654199/1/gate/puppet-openstack-beaker-centos-7/ef6abe1/ : SUCCESS in 12m 18s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/99/654199/1/gate/puppet-openstack-beaker-ubuntu-bionic/b1549a7/ : SUCCESS in 7m 51s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/99/654199/1/gate/build-openstack-releasenotes/20e4fff/html/ : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7a27d0926839889db4c98403cffd1f6c728dfb33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:30:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7853c55c412643432b0ac381ffd477ec7aad83eb","revisions":{"7853c55c412643432b0ac381ffd477ec7aad83eb":{"kind":"REWORK","_number":1,"created":"2019-04-21 06:45:42.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/99/654199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/99/654199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/99/654199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/99/654199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/99/654199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/99/654199/1"}}},"commit":{"parents":[{"commit":"feb4c0b839e97fbb9ec59d9930dcec3257de4b69","subject":"Fix Repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/feb4c0b839e97fbb9ec59d9930dcec3257de4b69"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:49.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:49.000000000","tz":480},"subject":"Move to opendev","message":"Move to opendev\n\nChange-Id: I9d06eeeb32b6d7d81312ecff71a848c58f903eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/7853c55c412643432b0ac381ffd477ec7aad83eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/7853c55c412643432b0ac381ffd477ec7aad83eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
