)]}'
{"id":"openstack%2Fpuppet-ovn~665807","triplet_id":"openstack%2Fpuppet-ovn~master~I0248fe0723451deee9a273e7e08b02675fcd2d7c","project":"openstack/puppet-ovn","branch":"master","hashtags":[],"change_id":"I0248fe0723451deee9a273e7e08b02675fcd2d7c","subject":"assert_private() params class","status":"ABANDONED","created":"2019-06-17 23:31:01.000000000","updated":"2020-09-18 22:57:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"56c05c2565c20f8ed077ffb874b984d2d3e37c1c","_number":665807,"virtual_id_number":665807,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-18 00:05:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 00:05:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"95fae327c0ae4f454f8852565fda673b0cc5f2f0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:31:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24cc03a226fa3039a2f438377023b3cd3c5bf996","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 00:05:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/07/665807/1/check/puppet-openstack-lint-ubuntu-bionic/35e9352/ : SUCCESS in 4m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/07/665807/1/check/puppet-openstack-syntax-5-ubuntu-bionic/a1f5152/ : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/07/665807/1/check/puppet-openstack-syntax-6-ubuntu-bionic/bcdcaf7/ : SUCCESS in 3m 58s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/07/665807/1/check/puppet-openstack-unit-5.5-centos-7/1882dd0/ : SUCCESS in 8m 55s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/07/665807/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/794727e/ : SUCCESS in 8m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/07/665807/1/check/puppet-openstack-unit-latest-ubuntu-bionic/383cb98/ : SUCCESS in 6m 32s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/07/665807/1/check/puppet-openstack-beaker-centos-7/38ebc4b/ : SUCCESS in 16m 55s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/07/665807/1/check/puppet-openstack-beaker-ubuntu-bionic/31703fd/ : FAILURE in 7m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56c05c2565c20f8ed077ffb874b984d2d3e37c1c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-18 22:57:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3cf8c288549cbf9d0fe4eb59faa0ea2347e97931","revisions":{"3cf8c288549cbf9d0fe4eb59faa0ea2347e97931":{"kind":"REWORK","_number":1,"created":"2019-06-17 23:31:01.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/07/665807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/07/665807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/07/665807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/07/665807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/07/665807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/07/665807/1"}}},"commit":{"parents":[{"commit":"58f95294581cd24aa667840aebae5c5bea099b0a","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/58f95294581cd24aa667840aebae5c5bea099b0a"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:30:51.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:30:51.000000000","tz":120},"subject":"assert_private() params class","message":"assert_private() params class\n\nChange-Id: I0248fe0723451deee9a273e7e08b02675fcd2d7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/3cf8c288549cbf9d0fe4eb59faa0ea2347e97931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/3cf8c288549cbf9d0fe4eb59faa0ea2347e97931"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
