)]}'
{"id":"openstack%2Fpuppet-ovn~672459","triplet_id":"openstack%2Fpuppet-ovn~master~Ic07e2a778bd148564e6dc337627235db2b611ca2","project":"openstack/puppet-ovn","branch":"master","topic":"prep-release-train-3","hashtags":[],"change_id":"Ic07e2a778bd148564e6dc337627235db2b611ca2","subject":"Prepare Train M3","status":"MERGED","created":"2019-07-24 08:20:28.000000000","updated":"2019-07-24 13:50:24.000000000","submitted":"2019-07-24 13:43:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672459-1563975781590-d6eb9d84","meta_rev_id":"bd00802868fe3297c537e3630abf0f70b447811f","_number":672459,"virtual_id_number":672459,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-24 13:43:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-07-24 13:32:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-07-24 13:32:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 13:32:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-24 13:43:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"093e33648fc129cac12d31f2529f4e899cc5adb4","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-24 08:20:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e15bb456ad61d5fc7df9f6eff301ec2b4939287","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 08:31:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/59/672459/1/check/build-openstack-puppet-tarball/2b0741c/ : SUCCESS in 2m 05s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/59/672459/1/check/puppet-openstack-lint-ubuntu-bionic/282437e/ : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"549b9227324710748537dd13a8ea965e75b0479b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-24 13:32:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e4f09c5b941dd28b7903b847bc7a0aa8f096dfca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:32:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2535f19a40e7dee3f69e580645fd9319b8f8b80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:43:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/59/672459/1/gate/build-openstack-puppet-tarball/404ff20/ : SUCCESS in 2m 50s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/59/672459/1/gate/puppet-openstack-lint-ubuntu-bionic/7a2e74e/ : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"fd4b68513afa73ee135c40887b938009817a8091","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:43:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c3f4cb793bc69ec942c48f0e567f1a6c2f3885d1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-24 13:50:24.000000000","message":"Patch Set 1:\n\njust realized it\u0027s m2, oh well","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef0e42f4dd601e9fc47de83a74df0c4ab8a1970e","revisions":{"ef0e42f4dd601e9fc47de83a74df0c4ab8a1970e":{"kind":"REWORK","_number":1,"created":"2019-07-24 08:20:28.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/59/672459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/59/672459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/59/672459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/59/672459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/59/672459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/59/672459/1"}}},"commit":{"parents":[{"commit":"58f95294581cd24aa667840aebae5c5bea099b0a","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/58f95294581cd24aa667840aebae5c5bea099b0a"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-24 08:19:39.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-24 08:19:39.000000000","tz":480},"subject":"Prepare Train M3","message":"Prepare Train M3\n\nUpdate the version metadata for Train milestone 3\n\nChange-Id: Ic07e2a778bd148564e6dc337627235db2b611ca2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/ef0e42f4dd601e9fc47de83a74df0c4ab8a1970e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/ef0e42f4dd601e9fc47de83a74df0c4ab8a1970e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
