)]}'
{"id":"openstack%2Fpuppet-ovn~673308","triplet_id":"openstack%2Fpuppet-ovn~stable%2Fstein~If8684927b17fc9f7ee4cca277b8b0ca9c4d8b5e1","project":"openstack/puppet-ovn","branch":"stable/stein","hashtags":[],"change_id":"If8684927b17fc9f7ee4cca277b8b0ca9c4d8b5e1","subject":"Add the ability to configure ovn-remote-probe-interval","status":"MERGED","created":"2019-07-29 15:09:38.000000000","updated":"2019-08-05 15:34:01.000000000","submitted":"2019-08-05 15:34:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673308-1565019241252-c5f855de","meta_rev_id":"284f1feba82a0002c03a7003b2f709cc5015aef7","_number":673308,"virtual_id_number":673308,"owner":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-05 15:34:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-02 15:49:54.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-05 13:46:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-05 13:46:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 15:49:54.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-05 13:46:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-05 15:34:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42904877746171c9862a949726a6a4447266e10b","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-07-29 15:09:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb9d34395a5405856cfe135942be40ed8fc521af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 15:29:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/08/673308/1/check/puppet-openstack-lint-ubuntu-bionic/926aea4/ : SUCCESS in 6m 43s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/08/673308/1/check/puppet-openstack-syntax-4-ubuntu-bionic/d1312e9/ : SUCCESS in 5m 54s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/08/673308/1/check/puppet-openstack-syntax-5-ubuntu-bionic/e679941/ : SUCCESS in 5m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/08/673308/1/check/puppet-openstack-syntax-6-ubuntu-bionic/b2ed432/ : SUCCESS in 5m 34s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/08/673308/1/check/puppet-openstack-unit-4.8-centos-7/e29d545/ : FAILURE in 8m 37s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/08/673308/1/check/puppet-openstack-unit-5.5-centos-7/38d714b/ : FAILURE in 7m 15s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/08/673308/1/check/puppet-openstack-beaker-centos-7/5ca54ea/ : SUCCESS in 12m 49s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/08/673308/1/check/puppet-openstack-beaker-ubuntu-bionic/713b4a7/ : FAILURE in 15m 22s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/08/673308/1/check/build-openstack-releasenotes/63612e0/html/ : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"1ed89e0259baa57ceafe0e5c3b6dabffb7dbbb19","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-02 15:49:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"631220ae31e53d659091317500205defe332d6ba","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-02 15:49:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9e9ee5d49e5d643a4a1600bfd3f8ed6bcaeb3715","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 16:15:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/08/673308/1/check/puppet-openstack-lint-ubuntu-bionic/68f9a7f/ : SUCCESS in 4m 50s\n- puppet-openstack-syntax-4-ubuntu-bionic https://logs.opendev.org/08/673308/1/check/puppet-openstack-syntax-4-ubuntu-bionic/bd21966/ : SUCCESS in 4m 14s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/08/673308/1/check/puppet-openstack-syntax-5-ubuntu-bionic/a16fd7d/ : SUCCESS in 5m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/08/673308/1/check/puppet-openstack-syntax-6-ubuntu-bionic/8a5cdfc/ : SUCCESS in 6m 51s\n- puppet-openstack-unit-4.8-centos-7 https://logs.opendev.org/08/673308/1/check/puppet-openstack-unit-4.8-centos-7/f7f1e78/ : SUCCESS in 13m 41s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/08/673308/1/check/puppet-openstack-unit-5.5-centos-7/92a9968/ : SUCCESS in 16m 09s\n- puppet-openstack-beaker-centos-7 https://logs.opendev.org/08/673308/1/check/puppet-openstack-beaker-centos-7/3a2c91e/ : SUCCESS in 19m 09s\n- puppet-openstack-beaker-ubuntu-bionic https://logs.opendev.org/08/673308/1/check/puppet-openstack-beaker-ubuntu-bionic/bb446d3/ : FAILURE in 8m 21s (non-voting)\n- build-openstack-releasenotes https://logs.opendev.org/08/673308/1/check/build-openstack-releasenotes/ae03527/html/ : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"bf82dbd8fce79c57ccd9734ab497196d137cc95b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 13:46:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db25c853db411c06ee9b05a1ebeab5873b244193","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 13:46:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffcf2793e745a4ae8cc3c6052ad9d94ad5eb126f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 14:21:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/08/673308/1/gate/puppet-openstack-lint-ubuntu-bionic/f188167/ : SUCCESS in 5m 37s\n- puppet-openstack-syntax-4-ubuntu-bionic https://logs.opendev.org/08/673308/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/aaa6092/ : SUCCESS in 7m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/08/673308/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/d14bd97/ : SUCCESS in 5m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/08/673308/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/f385fc6/ : SUCCESS in 5m 59s\n- puppet-openstack-unit-4.8-centos-7 https://logs.opendev.org/08/673308/1/gate/puppet-openstack-unit-4.8-centos-7/63793f5/ : SUCCESS in 12m 17s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/08/673308/1/gate/puppet-openstack-unit-5.5-centos-7/5991653/ : FAILURE in 8m 18s\n- puppet-openstack-beaker-centos-7 https://logs.opendev.org/08/673308/1/gate/puppet-openstack-beaker-centos-7/5f34f08/ : FAILURE in 25m 16s\n- puppet-openstack-beaker-ubuntu-bionic https://logs.opendev.org/08/673308/1/gate/puppet-openstack-beaker-ubuntu-bionic/b4c7c1f/ : FAILURE in 8m 54s (non-voting)\n- build-openstack-releasenotes https://logs.opendev.org/08/673308/1/gate/build-openstack-releasenotes/b5713f6/html/ : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"fd37fb4ff4945f563337cb5f3108416d76eb80f1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 14:23:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a1453284198b9bd665ffcfbec51699780f9bcda0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 14:59:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/08/673308/1/check/puppet-openstack-lint-ubuntu-bionic/7fec4a2/ : SUCCESS in 5m 22s\n- puppet-openstack-syntax-4-ubuntu-bionic https://logs.opendev.org/08/673308/1/check/puppet-openstack-syntax-4-ubuntu-bionic/47b6547/ : SUCCESS in 4m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/08/673308/1/check/puppet-openstack-syntax-5-ubuntu-bionic/5d4de43/ : SUCCESS in 5m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/08/673308/1/check/puppet-openstack-syntax-6-ubuntu-bionic/053ba16/ : SUCCESS in 5m 12s\n- puppet-openstack-unit-4.8-centos-7 https://logs.opendev.org/08/673308/1/check/puppet-openstack-unit-4.8-centos-7/8326823/ : SUCCESS in 9m 48s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/08/673308/1/check/puppet-openstack-unit-5.5-centos-7/8a76219/ : SUCCESS in 12m 41s\n- puppet-openstack-beaker-centos-7 https://logs.opendev.org/08/673308/1/check/puppet-openstack-beaker-centos-7/a36a258/ : SUCCESS in 12m 17s\n- puppet-openstack-beaker-ubuntu-bionic https://logs.opendev.org/08/673308/1/check/puppet-openstack-beaker-ubuntu-bionic/474d2e5/ : FAILURE in 9m 34s (non-voting)\n- build-openstack-releasenotes https://logs.opendev.org/08/673308/1/check/build-openstack-releasenotes/3155620/html/ : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"17b745f0cd925427fde8f33544e8f15b7c482483","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 14:59:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd4faabf25533865e89524a53dc6a1ac09f768ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 15:34:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/08/673308/1/gate/puppet-openstack-lint-ubuntu-bionic/2fe6304/ : SUCCESS in 5m 48s\n- puppet-openstack-syntax-4-ubuntu-bionic https://logs.opendev.org/08/673308/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/f6cc5ee/ : SUCCESS in 5m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/08/673308/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/799846a/ : SUCCESS in 6m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/08/673308/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/8fc811f/ : SUCCESS in 6m 44s\n- puppet-openstack-unit-4.8-centos-7 https://logs.opendev.org/08/673308/1/gate/puppet-openstack-unit-4.8-centos-7/cafe2b7/ : SUCCESS in 10m 38s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/08/673308/1/gate/puppet-openstack-unit-5.5-centos-7/66b6954/ : SUCCESS in 11m 28s\n- puppet-openstack-beaker-centos-7 https://logs.opendev.org/08/673308/1/gate/puppet-openstack-beaker-centos-7/741ece2/ : SUCCESS in 16m 09s\n- puppet-openstack-beaker-ubuntu-bionic https://logs.opendev.org/08/673308/1/gate/puppet-openstack-beaker-ubuntu-bionic/0b12d50/ : FAILURE in 7m 39s (non-voting)\n- build-openstack-releasenotes https://logs.opendev.org/08/673308/1/gate/build-openstack-releasenotes/ab99ea6/html/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"284f1feba82a0002c03a7003b2f709cc5015aef7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 15:34:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d401a869fd5ba264533f3046c8de5e585b48be6","revisions":{"2d401a869fd5ba264533f3046c8de5e585b48be6":{"kind":"REWORK","_number":1,"created":"2019-07-29 15:09:38.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/08/673308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/08/673308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/08/673308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/08/673308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/08/673308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/08/673308/1"}}},"commit":{"parents":[{"commit":"69b547995f5f07a7c55f3a0a2c524b21b313d1ef","subject":"Add DPDK support for OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/69b547995f5f07a7c55f3a0a2c524b21b313d1ef"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-07-15 15:15:26.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-07-29 15:09:36.000000000","tz":120},"subject":"Add the ability to configure ovn-remote-probe-interval","message":"Add the ability to configure ovn-remote-probe-interval\n\nThis parameter sets inactive probe interval of the JSON\nsession from ovn-controller to the OVN SB database.\nBy default this it is 5s which not be sufficient in\nloaded systems or during high control-plane activity spikes,\nleading to unnecessary reconnections to OVSDB server.\nNow it is extended by default to 1 min and it\nis configurable.\n\nChange-Id: If8684927b17fc9f7ee4cca277b8b0ca9c4d8b5e1\nPartial-Bug: #1836604\n(cherry picked from commit 058b8e1cbd595faf05c513af2acd71fcbafc0fcd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/2d401a869fd5ba264533f3046c8de5e585b48be6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/2d401a869fd5ba264533f3046c8de5e585b48be6"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
