)]}'
{"id":"openstack%2Fpuppet-ovn~680270","triplet_id":"openstack%2Fpuppet-ovn~master~Ib5275b70bfbad82d9d05c820ddb5c02246f9ba83","project":"openstack/puppet-ovn","branch":"master","topic":"prep-release-train-4","hashtags":[],"change_id":"Ib5275b70bfbad82d9d05c820ddb5c02246f9ba83","subject":"Prepare Train M4","status":"MERGED","created":"2019-09-05 01:06:02.000000000","updated":"2019-09-06 14:10:55.000000000","submitted":"2019-09-06 14:10:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680270-1567779055425-e4e8de68","meta_rev_id":"5d182c867e27b2df6439d812231f5c57581b0c1d","_number":680270,"virtual_id_number":680270,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-09-06 14:10:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-09-06 14:02:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-09-06 14:02:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 14:02:29.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-06 14:10:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73d9f7b5d4b0e2eeede3858c568f539688d834a0","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-09-05 01:06:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43baeee2dbfdf7aff9ede4c4dabc143d7186cecf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 01:50:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/08d10c6a159142abbeb88a550629fdb2 : SUCCESS in 2m 39s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d35f3480eded48f3bd4c8e4cdf7e42ff : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"39cab6d64afa120db22bfb52cf4a354829297925","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-06 14:02:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d36ab448fafff467691608e47b19565b8b4e39d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:03:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e218408f9178186a030bec4c6112da5f02d0308","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:10:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0982431e34cdc879c4bb4dbc8569d910d9bdc3b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:10:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/1a27f9c2bb364be88f661b61f1783b0c : SUCCESS in 2m 22s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9153de07b964d7599e3f78d8a77057e : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ce7790a5f6267bf58bbf0831c81bacb573fb95e","revisions":{"3ce7790a5f6267bf58bbf0831c81bacb573fb95e":{"kind":"REWORK","_number":1,"created":"2019-09-05 01:06:02.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/70/680270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/70/680270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/70/680270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/70/680270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/70/680270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/70/680270/1"}}},"commit":{"parents":[{"commit":"5df782dd3fde83a00dd1dbf1226fa2c8a2ecb902","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/5df782dd3fde83a00dd1dbf1226fa2c8a2ecb902"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-05 01:04:50.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-05 01:04:50.000000000","tz":480},"subject":"Prepare Train M4","message":"Prepare Train M4\n\nUpdate the version metadata for Train milestone 4\n\nChange-Id: Ib5275b70bfbad82d9d05c820ddb5c02246f9ba83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/3ce7790a5f6267bf58bbf0831c81bacb573fb95e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/3ce7790a5f6267bf58bbf0831c81bacb573fb95e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
