)]}'
{"id":"openstack%2Fpuppet-ovn~683291","triplet_id":"openstack%2Fpuppet-ovn~master~I98c7b16c6051e174e515bfa9a81a3fb55267aacf","project":"openstack/puppet-ovn","branch":"master","topic":"feature/add_ovn_openflow_probe_interval","hashtags":[],"change_id":"I98c7b16c6051e174e515bfa9a81a3fb55267aacf","subject":"Add posibilities to set ovn_openflow_probe_interval for controller","status":"MERGED","created":"2019-09-20 06:24:50.000000000","updated":"2019-09-21 01:24:39.000000000","submitted":"2019-09-21 01:23:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683291-1569029011809-0f18a21d","meta_rev_id":"f68a36c871117b277a344a8192f9286da89b67e1","_number":683291,"virtual_id_number":683291,"owner":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-09-20 13:12:55.000000000","_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":2,"date":"2019-09-21 01:23:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-09-21 01:03:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-09-20 13:14:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2019-09-20 14:23:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":1,"date":"2019-09-20 15:34:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-09-21 01:03:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 08:31:50.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-09-20 08:42:26.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-20 13:14:20.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2019-09-20 14:23:59.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-20 15:34:32.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-09-21 01:03:39.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-09-21 01:23:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae6eb2d5bb197db5f6c8e5aac652e9055bb0103b","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-09-20 06:24:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a08cbc0a59ba6962845be20526e03e99a7d30924","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 07:15:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35ec3ab421d44ead9a1d8b3a056ad590 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40e9ce08908841818e28c1babe0d77e8 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adc7d8ed3c9e46f78fa00eeeff5102b1 : SUCCESS in 4m 10s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/812e313fcfd64d089bd0c376eda2c1e5 : SUCCESS in 9m 43s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/611085c7be704ee1afc14aca0dea147a : SUCCESS in 7m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ec3ad6d599e42cca3d07f743d6e59e5 : SUCCESS in 6m 23s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1726d512346b4716a2bc532171b76b70 : SUCCESS in 13m 22s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/674ee224fe934cc6939097f29c94fee1 : FAILURE in 10m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/887c88f233e54efe994518be8c748f5b : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"7312463e245b36a55d3846d2fbaa86cc7e2428fe","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-09-20 08:28:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6ec21f270334da2eded850a9d73e485ae7f03ab5","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-09-20 08:31:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93e51e90a60e90741464649bf296ccb5a594c0df","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-09-20 08:36:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0204b9e89ccf4b83e2b6cf126becc26330d294e3","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-09-20 08:42:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9aa0cd963bd0bfca8d5d234763ab6b340785a475","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 08:55:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0083dce1a7a84916a67fb5857d8aca8a : SUCCESS in 5m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d5167c531fa4ed3b1a9024f234f5b16 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b33eda79b434f7d8f026fe9170f2a9b : SUCCESS in 3m 50s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a6dbc7590e5e4df4ae90f15d4b3e6062 : SUCCESS in 8m 15s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e28c5d5b99114fac8377de5b78f17eee : SUCCESS in 6m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04e1b062427f4ac099dcc71c020014f0 : SUCCESS in 6m 56s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9562daaa8c2c40f2af0bc01cd67cb663 : SUCCESS in 12m 18s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aac2029d965245e6a6b2f7cb4d0f50d8 : FAILURE in 6m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1d252e8fb034321bfb4ba1e9cd2dfb3 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":2},{"id":"12fed69efb92968f530f852a489228a3d75ae647","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-09-20 12:09:14.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"af2adc7d2e63fc4209a1aab159eb0ee8edb0e63b","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-09-20 13:11:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c708b875645bef85b89db25c72107dfeb8c6a92a","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-09-20 13:12:55.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a04a093d0ebe48361a6c06c30aaaec6c93303ee0","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-09-20 13:14:20.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"becb439cc878d6af445a473afb61b2b88f4c8deb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 13:49:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d927d1d7cf134d578aa98e904e404fe8 : SUCCESS in 4m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e06972446944644948c88107a2fcfb7 : SUCCESS in 4m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/712641ca2aa24f5992f3cb26f760f4b9 : SUCCESS in 3m 37s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f6c6f6264e6a4ff88e7d5c5c1fd3411f : SUCCESS in 10m 12s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12f96ce0b6454c05bf0ca4049c5e5214 : SUCCESS in 12m 09s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c215b38d658444897faeec0b28773aa : SUCCESS in 5m 42s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4be39c3b29c44721ad945a38b90f60da : SUCCESS in 11m 39s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e86f504cca5047549ee31c5d7efcf11b : FAILURE in 8m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd98e8d194a24ccaacd1d33635d5c060 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":3},{"id":"edcccfa68ef9e9c33e7062874cb04580488f0f70","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-20 14:23:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ad49635e067589a95365498163a1cea99bb05088","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-09-20 15:34:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"193345af01590ce127ee0d3d65e498a72a05a819","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-09-21 01:03:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a055a0b253dfd91772b8c188569e7b35861fe0a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 01:03:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"adfe4f0628a8ef8908eea557a6b2286aaa747807","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 01:23:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6aa363456e7f48de9a46038ca55d4791 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6ba9b0862234cc7b66cd80d8868de5b : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87771950b42841cc9a9bb9e970ea7dc3 : SUCCESS in 4m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ffe520eeba0e4e0c9e5c1390a300395e : SUCCESS in 8m 32s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b677e29924f84dbb89e5144ec259967c : SUCCESS in 6m 18s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/94c8060efbd64f919ad399eadef1c8fb : SUCCESS in 12m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77c07964db5d4a63bfeefac28cbf4f56 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":3},{"id":"9160e844a8f5c2356f0191f85642fa5ff70857f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 01:23:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0638ef2f90b6f15bea92cf22355d5aa094f0463c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 01:24:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e5a6e8d3a89469c8175e73b353e15c3 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d1331a6254bc0d86a4fb6be78b6bc8dbe97ec021","revisions":{"49f810b0e3f58d6dfc64228b398e583e93040693":{"kind":"REWORK","_number":1,"created":"2019-09-20 06:24:50.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/91/683291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/91/683291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/91/683291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/91/683291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/91/683291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/91/683291/1"}}},"commit":{"parents":[{"commit":"3ce7790a5f6267bf58bbf0831c81bacb573fb95e","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/3ce7790a5f6267bf58bbf0831c81bacb573fb95e"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-09-20 06:17:04.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-09-20 06:17:04.000000000","tz":120},"subject":"Add posibilities to set ovn_openflow_probe_interval for controller","message":"Add posibilities to set ovn_openflow_probe_interval for controller\n\nChange-Id: I98c7b16c6051e174e515bfa9a81a3fb55267aacf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/49f810b0e3f58d6dfc64228b398e583e93040693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/49f810b0e3f58d6dfc64228b398e583e93040693"}]},"branch":"refs/heads/master"},"8dd1c1cd3bff8812401106581870e7fdf01e23fd":{"kind":"REWORK","_number":2,"created":"2019-09-20 08:36:58.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/91/683291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/91/683291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/91/683291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/91/683291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/91/683291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/91/683291/2"}}},"commit":{"parents":[{"commit":"3ce7790a5f6267bf58bbf0831c81bacb573fb95e","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/3ce7790a5f6267bf58bbf0831c81bacb573fb95e"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-09-20 06:17:04.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-09-20 08:36:46.000000000","tz":120},"subject":"Add posibilities to set ovn_openflow_probe_interval for controller","message":"Add posibilities to set ovn_openflow_probe_interval for controller\n\nChange-Id: I98c7b16c6051e174e515bfa9a81a3fb55267aacf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/8dd1c1cd3bff8812401106581870e7fdf01e23fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/8dd1c1cd3bff8812401106581870e7fdf01e23fd"}]},"branch":"refs/heads/master"},"d1331a6254bc0d86a4fb6be78b6bc8dbe97ec021":{"kind":"REWORK","_number":3,"created":"2019-09-20 13:11:09.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/91/683291/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/91/683291/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/91/683291/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/91/683291/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/91/683291/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/91/683291/3"}}},"commit":{"parents":[{"commit":"3ce7790a5f6267bf58bbf0831c81bacb573fb95e","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/3ce7790a5f6267bf58bbf0831c81bacb573fb95e"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-09-20 06:17:04.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-09-20 13:10:15.000000000","tz":120},"subject":"Add posibilities to set ovn_openflow_probe_interval for controller","message":"Add posibilities to set ovn_openflow_probe_interval for controller\n\nChange-Id: I98c7b16c6051e174e515bfa9a81a3fb55267aacf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/d1331a6254bc0d86a4fb6be78b6bc8dbe97ec021"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/d1331a6254bc0d86a4fb6be78b6bc8dbe97ec021"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
