)]}'
{"id":"openstack%2Fpuppet-ovn~690020","triplet_id":"openstack%2Fpuppet-ovn~stable%2Fstein~I98c7b16c6051e174e515bfa9a81a3fb55267aacf","project":"openstack/puppet-ovn","branch":"stable/stein","topic":"ovn_openflow_probe_interval","hashtags":[],"change_id":"I98c7b16c6051e174e515bfa9a81a3fb55267aacf","subject":"Add posibilities to set ovn_openflow_probe_interval for controller","status":"MERGED","created":"2019-10-22 09:24:09.000000000","updated":"2019-10-24 20:59:38.000000000","submitted":"2019-10-24 20:57:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690020-1571950665598-edb63827","meta_rev_id":"c8efe1e65839b5e70293db17ba5a4f48e7661129","_number":690020,"virtual_id_number":690020,"owner":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-24 20:57:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-10-24 07:25:41.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-10-24 20:40:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-10-24 20:40:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 07:25:41.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-24 20:40:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-24 20:57:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"481e661abc65eb479fd5e6b8f7f14f23c3cc2219","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-10-22 09:24:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"988b6c166470b85a16454cf4a6e534ca045ea542","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 09:40:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4ebbe748313499c9e294d43cab8853a : SUCCESS in 3m 43s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b825975d4e484f36a12e8ae0328f1851 : SUCCESS in 4m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03d5ca47f14545d89a3b618787f73f34 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1343742996649ebb07545482e533105 : SUCCESS in 4m 02s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/339f4e071dd74b7f8dfa0d9374393cbc : SUCCESS in 10m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a7bd75b3a90345a8b86a437306b5a8b9 : SUCCESS in 8m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/dcd1d03358a64306bbecb15b5c0ba592 : SUCCESS in 10m 52s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/458f5fc3a1b34a169dc2210ea5e40365 : FAILURE in 6m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/220a7fdeccfa444d8f1dc7cf881c7300 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c9a217b8fc40a1c421675e55eb308c33133fb1f2","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-10-22 10:02:59.000000000","message":"Topic set to ovn_openflow_probe_interval","accounts_in_message":[],"_revision_number":1},{"id":"1a849780cbed8d14fa151e1cd44e1406deb653b3","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-24 07:25:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4dbb14072c822889bd05fa3edbef3c5376872b3d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-24 20:40:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e3f2a4ebbb361f6a64c70f34c6dba241160ff4e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 20:40:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1df81fcac9ca4018b7c0c8f520209e96894eeff1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 20:57:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4011ffdbf3f4d17bad87ff6a98ad5ab : SUCCESS in 4m 49s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fbb668497bc4163b239f36535cd86bd : SUCCESS in 4m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d37e120e6910427bb4d2a7212fdc4cd4 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebdc9e51df98479eb4253243c0deeda9 : SUCCESS in 5m 06s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/2c611f5ffe304840a3ecd62dc055b7f6 : SUCCESS in 10m 31s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/20fea139dbd6458dbfe97c3f846c6b7e : SUCCESS in 9m 10s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5d70ace8a79a46b4a449174c4f10df8a : SUCCESS in 11m 31s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7177354cd63d4013bc15609658947340 : SUCCESS in 8m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b7cd9ca8d6444afa50d74225df13f6b : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"03e8ca4c0a98278ca2be260dd19fada01ff77f32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 20:57:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c8efe1e65839b5e70293db17ba5a4f48e7661129","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 20:59:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab944c74a2ce4e0bb35411befb7d2bd6 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5b784df19a61ec423d09c2549cefca00c586314","revisions":{"a5b784df19a61ec423d09c2549cefca00c586314":{"kind":"REWORK","_number":1,"created":"2019-10-22 09:24:09.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/20/690020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/20/690020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/20/690020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/20/690020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/20/690020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/20/690020/1"}}},"commit":{"parents":[{"commit":"2d401a869fd5ba264533f3046c8de5e585b48be6","subject":"Add the ability to configure ovn-remote-probe-interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/2d401a869fd5ba264533f3046c8de5e585b48be6"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-09-20 06:17:04.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-10-22 09:23:48.000000000","tz":120},"subject":"Add posibilities to set ovn_openflow_probe_interval for controller","message":"Add posibilities to set ovn_openflow_probe_interval for controller\n\nChange-Id: I98c7b16c6051e174e515bfa9a81a3fb55267aacf\n(cherry picked from commit d1331a6254bc0d86a4fb6be78b6bc8dbe97ec021)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/a5b784df19a61ec423d09c2549cefca00c586314"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/a5b784df19a61ec423d09c2549cefca00c586314"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
