)]}'
{"id":"openstack%2Fpuppet-ovn~690027","triplet_id":"openstack%2Fpuppet-ovn~stable%2Frocky~I98c7b16c6051e174e515bfa9a81a3fb55267aacf","project":"openstack/puppet-ovn","branch":"stable/rocky","topic":"ovn_openflow_probe_interval","hashtags":[],"change_id":"I98c7b16c6051e174e515bfa9a81a3fb55267aacf","subject":"Add posibilities to set ovn_openflow_probe_interval for controller","status":"MERGED","created":"2019-10-22 09:49:03.000000000","updated":"2019-10-22 14:11:29.000000000","submitted":"2019-10-22 14:10:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690027-1571753410473-fe29816b","meta_rev_id":"22e75ecd1932252854129d43117a4c140ac22489","_number":690027,"virtual_id_number":690027,"owner":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-22 14:10:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-10-22 13:52:26.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-10-22 13:52:26.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 13:52:26.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-22 14:10:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f093b944955abcaeafa5318254e49d54e58d6b6d","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-10-22 09:49:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"132233aae604508796e85a233b93b05545c6c078","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-10-22 10:02:52.000000000","message":"Topic changed from fix/back_rocky to ovn_openflow_probe_interval","accounts_in_message":[],"_revision_number":1},{"id":"04d7aef29f3af2ff04db727fa8ecfc5d46401a55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 10:05:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/cdd19d0d2210406c85f27020b7a36ea9 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/1958f14f6c5d4d12b3ea2c8fc8204832 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/4d44472298c04ca592836976ebd35ccb : SUCCESS in 4m 52s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/c99ab72922eb412e9421c36f5d39e0ac : SUCCESS in 7m 31s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9420e93120e54189b34171d32e7e8fcc : SUCCESS in 11m 04s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a73872a5e479447d84327f5acce7267c : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7947f67fc5224f6c8baebb6ecad429b8 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"55a6a6d8803725c9b9a29b67049524b90e9423b5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-22 13:52:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"036a267123f43579c495e81698eeae0adef015b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 13:52:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5704231750682bf654dc2a756acda1ec269f083","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 14:10:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ec0a9507df4bb7920f296453820aae81f38e5bd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 14:10:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/a355ddd062e64ad7ad3f3f1ced484cb0 : SUCCESS in 5m 08s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/54ce092ed7074c2ca14ff4ab09cbceef : SUCCESS in 4m 18s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/eabe9f2fde444a388536c06fd170a9a5 : SUCCESS in 4m 32s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/0f038ddb4e544af2a5f3f47989e32d0d : SUCCESS in 9m 58s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/384667d0222a48c4b55b1610e769a7a5 : SUCCESS in 12m 26s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03de5e714c4940889dd70e8db0a615f7 : SUCCESS in 9m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b79b2e2e754e488687472de94df6a522 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"22e75ecd1932252854129d43117a4c140ac22489","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 14:11:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c64facb714da4c789360330b11d59470 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9fdaad0e769107247c15de7f42308aa7f3c145ed","revisions":{"9fdaad0e769107247c15de7f42308aa7f3c145ed":{"kind":"REWORK","_number":1,"created":"2019-10-22 09:49:03.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/27/690027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/27/690027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/27/690027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/27/690027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/27/690027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/27/690027/1"}}},"commit":{"parents":[{"commit":"f68e399cf11f2b3ae6bb74babf04e76e9d0503c2","subject":"Add the ability to configure ovn-remote-probe-interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/f68e399cf11f2b3ae6bb74babf04e76e9d0503c2"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-09-20 06:17:04.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-10-22 09:48:50.000000000","tz":120},"subject":"Add posibilities to set ovn_openflow_probe_interval for controller","message":"Add posibilities to set ovn_openflow_probe_interval for controller\n\nChange-Id: I98c7b16c6051e174e515bfa9a81a3fb55267aacf\n(cherry picked from commit d1331a6254bc0d86a4fb6be78b6bc8dbe97ec021)\n(cherry picked from commit a5b784df19a61ec423d09c2549cefca00c586314)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/9fdaad0e769107247c15de7f42308aa7f3c145ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/9fdaad0e769107247c15de7f42308aa7f3c145ed"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
