)]}'
{"id":"openstack%2Fpuppet-ovn~690033","triplet_id":"openstack%2Fpuppet-ovn~stable%2Fqueens~I98c7b16c6051e174e515bfa9a81a3fb55267aacf","project":"openstack/puppet-ovn","branch":"stable/queens","topic":"fix/ovn_openflow_probe_interval","hashtags":[],"change_id":"I98c7b16c6051e174e515bfa9a81a3fb55267aacf","subject":"Add posibilities to set ovn_openflow_probe_interval for controller","status":"MERGED","created":"2019-10-22 10:09:11.000000000","updated":"2019-10-25 12:46:52.000000000","submitted":"2019-10-25 12:44:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690033-1572007481009-5cb63f8a","meta_rev_id":"ecb0d008e8e17b8df42179abb6d1cccaf61ff451","_number":690033,"virtual_id_number":690033,"owner":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 12:44:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-10-25 12:17:26.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-10-24 20:40:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-10-25 12:29:58.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 20:40:08.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-25 12:29:58.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-25 12:44:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0eb4f763ff9ab1563f96d14ea30cd9d61db30c0","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-10-22 10:09:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bfe88e51fcf05b7338005ed616ef35587c50fd0","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-10-22 10:12:03.000000000","message":"Topic changed from fix/ovn_openflow_probe_interval to ovn_openflow_probe_interval","accounts_in_message":[],"_revision_number":1},{"id":"76c689e0919285c9ef6941110b0c018663102802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 10:24:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/a00a29a59cec47268ff39d2aade550bc : FAILURE in 5m 16s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/5d8e2e78e63c4d9783f6d5be4eb63024 : SUCCESS in 3m 40s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/495f924e9eaf467f8c596e2c86bacebc : SUCCESS in 8m 05s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/18df301f89dd4fb6bd961489a240cce6 : SUCCESS in 8m 33s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e5a528a5bc5843e6a26d2bb23744056a : SUCCESS in 11m 44s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/869480496b5d4ca8832109c497923ff8 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/381be67d67294ebbb36c1d0ca2929b2c : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"209699fcd32694cd39b4ebeaae7264f029defd15","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-10-22 14:57:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"65425d167ee53ae6e6eea75dc40a5078812f44b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 15:14:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/b9872d7810c74758af8160fe806e4598 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/9c815b2394ef415da91b80e8e4abcd70 : SUCCESS in 4m 58s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/6d01443199f34966bdd29f9c05dc9ecb : SUCCESS in 8m 33s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/dce9a4be162741199d6d3e66d0d4f5e0 : SUCCESS in 8m 54s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e1185117e28f49e8bb51bb246e8517b1 : SUCCESS in 14m 52s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/f8c00fe4328d487c9c1704ee3a2d7afd : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a5aa17a0c014ff9a62a07ede55640cd : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":2},{"id":"99c829aa5c30e7d4b22392988c0126e87dc1ab8d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-24 20:40:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f8385ffda55ac79998dffec12968a1730eb18371","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-25 12:17:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5541846f687836c8298944740127be1d4a75efb0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-25 12:29:58.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a3e9e06c50d6826ecab183d9ce839cc5ef480398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 12:30:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"846d13d242e5f423fb81c0c6e7972ec0722bff3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 12:44:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/51867bc6edc04ee792fff5a7e5b3a10f : SUCCESS in 6m 30s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/7c07d2157d8041168e0814f0d4bcd509 : SUCCESS in 5m 18s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/203807eb78c947e5b839db571d9b775a : SUCCESS in 10m 01s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/63ea57504c904f0bb48f7c27e427786a : SUCCESS in 10m 27s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/066db9cf6a51429a887b5929f8ca8a1d : SUCCESS in 12m 37s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/14574890bad54050a5401ae5a8103557 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15f1456a222b4f719bd578ea29ffa2c2 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":2},{"id":"6133b6a11071c60d76f19dc44d6d5d6b42324816","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 12:44:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5d932f718644608a6848d8380ddd4104d1106b61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 12:46:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c2601d284884bac973588b89504ff7d : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"41c885cf4c558d6e44a3708c02d93acc34b348da","revisions":{"3aa0d380091b6b0c8ceca13b5998c873f9b6957e":{"kind":"REWORK","_number":1,"created":"2019-10-22 10:09:11.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/33/690033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/33/690033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/690033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/690033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/690033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/33/690033/1"}}},"commit":{"parents":[{"commit":"56bcc3be9608367fb8c745d96296a194c439118f","subject":"Add the ability to configure ovn-remote-probe-interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/56bcc3be9608367fb8c745d96296a194c439118f"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-09-20 06:17:04.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-10-22 10:08:57.000000000","tz":120},"subject":"Add posibilities to set ovn_openflow_probe_interval for controller","message":"Add posibilities to set ovn_openflow_probe_interval for controller\n\nChange-Id: I98c7b16c6051e174e515bfa9a81a3fb55267aacf\n(cherry picked from commit d1331a6254bc0d86a4fb6be78b6bc8dbe97ec021)\n(cherry picked from commit a5b784df19a61ec423d09c2549cefca00c586314)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/3aa0d380091b6b0c8ceca13b5998c873f9b6957e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/3aa0d380091b6b0c8ceca13b5998c873f9b6957e"}]},"branch":"refs/heads/stable/queens"},"41c885cf4c558d6e44a3708c02d93acc34b348da":{"kind":"REWORK","_number":2,"created":"2019-10-22 14:57:22.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/33/690033/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/33/690033/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/690033/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/690033/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/690033/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/33/690033/2"}}},"commit":{"parents":[{"commit":"56bcc3be9608367fb8c745d96296a194c439118f","subject":"Add the ability to configure ovn-remote-probe-interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/56bcc3be9608367fb8c745d96296a194c439118f"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-09-20 06:17:04.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-10-22 14:57:07.000000000","tz":120},"subject":"Add posibilities to set ovn_openflow_probe_interval for controller","message":"Add posibilities to set ovn_openflow_probe_interval for controller\n\nChange-Id: I98c7b16c6051e174e515bfa9a81a3fb55267aacf\n(cherry picked from commit d1331a6254bc0d86a4fb6be78b6bc8dbe97ec021)\n(cherry picked from commit a5b784df19a61ec423d09c2549cefca00c586314)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/41c885cf4c558d6e44a3708c02d93acc34b348da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/41c885cf4c558d6e44a3708c02d93acc34b348da"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
