)]}'
{"id":"openstack%2Fpuppet-ovn~700085","triplet_id":"openstack%2Fpuppet-ovn~master~I740c3cf6ac907a0d6854680ee89b7ca60f3407d6","project":"openstack/puppet-ovn","branch":"master","hashtags":[],"change_id":"I740c3cf6ac907a0d6854680ee89b7ca60f3407d6","subject":"Add clustering support for ovsdb","status":"MERGED","created":"2019-12-19 23:15:22.000000000","updated":"2020-01-07 19:56:25.000000000","submitted":"2020-01-07 19:56:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700085-1578426985452-893a25b6","meta_rev_id":"2de9dc9632bd86266a1573b75cfd284fdc62db7e","_number":700085,"virtual_id_number":700085,"owner":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-07 19:56:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-01-07 19:39:14.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-01-07 14:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-01-07 19:39:14.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-07 14:35:39.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-01-07 19:39:14.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-01-07 19:56:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e8440b8c7ca5578d7315295e04168baf84a8d9f","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2019-12-19 23:15:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15e35a30dc3a670520a7a695940f11bd8968c816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 23:32:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/555a2c43176c4465a0b73aaf5ec57f0a : FAILURE in 7m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35ce4a6ecb614d16b80415a6265af0cc : SUCCESS in 6m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c2b145e9f6745a1acc12a8fab7437b7 : SUCCESS in 5m 02s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4818707c264b414891232d66441bda32 : SUCCESS in 8m 49s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f141efed5164e1491a3090c61869d5d : SUCCESS in 6m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2b1b7af7723414399f667ff979113d3 : SUCCESS in 7m 41s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/56aff97dbb5c468b87fd07450e77f0fe : SUCCESS in 11m 55s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1697ee944de64f14aa7fcf0e6aae8908 : SUCCESS in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"589c37abdada6bf84e80bb25f719ee36a0511599","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2019-12-20 03:42:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6b1f08535449707a2d061fcc81d5696b832408ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 03:56:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb1999e0b45240ef82882dfec1afc44f : FAILURE in 4m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47cd3771d3d148438cd5b0e7dcb64a98 : SUCCESS in 4m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c1aee5431ff42ab8dc6c779ea03f9ab : SUCCESS in 4m 42s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9f1ad58436494e2fa528b8e337dc50b6 : SUCCESS in 9m 01s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c68a3b2094da4b53adde4cb4ad38529c : SUCCESS in 6m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb4d2a6c97ea4ab9ad91efbc43bff74d : SUCCESS in 7m 14s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/753e7893ff564700af4c5c5ae82bb72b : SUCCESS in 12m 33s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbeaf246a0224645ad50a6b17db9bfbc : SUCCESS in 9m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4dead93756184fde5a38fabbb7a299489fa6002e","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2019-12-23 00:01:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"63b6472f95dcc12e9cf1ac760ad12c3bcd39ac21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 00:42:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfc579ff04fa4d01bafb52eee0e60934 : SUCCESS in 6m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab109cc45bf34a14bde9f878ea1a7d6b : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57ac4e8fe6eb4200b87ae9c7c8537c24 : SUCCESS in 6m 55s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/fbd16953afb1403a8f62b2fb40afbb78 : SUCCESS in 9m 18s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec89029823dd4b4891c43bbc25f4855e : SUCCESS in 11m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/578ddabdc6a14a6fbd9ea68431f88aac : SUCCESS in 7m 52s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c3c5f7f4ca5e48eaaa251c6231ff119b : SUCCESS in 13m 32s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff4901e046a6435bbd28d30f2f931b09 : SUCCESS in 11m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"179becdb5dee1284cdb084ea1ad5c964fc183aeb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-01-06 21:49:26.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ac98b4f396a4dde675c68ce87a139c2632b2b215","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2020-01-07 05:27:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"48f91abb561bdfa4eff32cd8fbe89c935658108e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 05:44:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c294f633b9a74bcfa01a0129284a5a1e : SUCCESS in 5m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2dcfa2cc72844fae8fcb08f3717f5d5f : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5853089414f945d5945981d9398b0b8a : SUCCESS in 3m 57s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/068a3d94168f448fb14f51179fdf96a5 : SUCCESS in 7m 51s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16fe7face4304c41bc761f938f262343 : SUCCESS in 6m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9d59f1b2f724c41bbd230f67d6b07fd : SUCCESS in 6m 42s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/36b0ff02bc24431abb03de8f88563e49 : SUCCESS in 11m 40s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e9f4177f7654a11bb4b99091f13e496 : SUCCESS in 8m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6e9a7b022e0c37e3b7bb41e18a6d25a315e12eb7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-01-07 14:35:39.000000000","message":"Patch Set 4: Code-Review+2\n\ntests would be nice","accounts_in_message":[],"_revision_number":4},{"id":"0f5e771cf1326b8035a95d5afd0d9bc441386105","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-01-07 19:39:14.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d7d47c58c49ccc13420befb76d9ec70b6d271ecd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 19:39:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bd94a2a9cf6bee02885213f714143fda257b0dbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 19:56:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2de9dc9632bd86266a1573b75cfd284fdc62db7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 19:56:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00cd9a6c0de14ddfb8caf42436c6b14d : SUCCESS in 5m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f371a8f7c8c4e848531032519638946 : SUCCESS in 4m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f453681f9a414dac91af32e18c75d05f : SUCCESS in 5m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/221c02dc2ff2491b8615862d31443684 : SUCCESS in 12m 38s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90f8377cbf7d4813b71a1337d15f0dc8 : SUCCESS in 7m 55s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d877cded743542aa8aef76833205803b : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"82cadd3a280e60e12f6720b3842b9e1ffbd7c9b2","revisions":{"a5197ba47905cd5568b34bd8c91669cda0badc79":{"kind":"REWORK","_number":1,"created":"2019-12-19 23:15:22.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/85/700085/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/85/700085/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/1"}}},"commit":{"parents":[{"commit":"f4f596d4a1def4e9795de91c4b56102d18cf5d59","subject":"Add Debian support for initialising DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/f4f596d4a1def4e9795de91c4b56102d18cf5d59"}]}],"author":{"name":"Jake Yip","email":"jake.yip@unimelb.edu.au","date":"2019-12-18 06:23:55.000000000","tz":660},"committer":{"name":"Jake Yip","email":"jake.yip@unimelb.edu.au","date":"2019-12-19 23:13:31.000000000","tz":660},"subject":"Add clustering support for ovsdb","message":"Add clustering support for ovsdb\n\nChange-Id: I740c3cf6ac907a0d6854680ee89b7ca60f3407d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/a5197ba47905cd5568b34bd8c91669cda0badc79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/a5197ba47905cd5568b34bd8c91669cda0badc79"}]},"branch":"refs/heads/master"},"74cf36da5ac69cce2b8a17fd40fa279b83b68c84":{"kind":"REWORK","_number":2,"created":"2019-12-20 03:42:23.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/85/700085/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/85/700085/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/2"}}},"commit":{"parents":[{"commit":"77bc7d77dad5d95c6e87ba086de3d6e94f0014ac","subject":"Add Debian support for initialising DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/77bc7d77dad5d95c6e87ba086de3d6e94f0014ac"}]}],"author":{"name":"Jake Yip","email":"jake.yip@unimelb.edu.au","date":"2019-12-18 06:23:55.000000000","tz":660},"committer":{"name":"Jake Yip","email":"jake.yip@unimelb.edu.au","date":"2019-12-20 03:42:09.000000000","tz":660},"subject":"Add clustering support for ovsdb","message":"Add clustering support for ovsdb\n\nChange-Id: I740c3cf6ac907a0d6854680ee89b7ca60f3407d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/74cf36da5ac69cce2b8a17fd40fa279b83b68c84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/74cf36da5ac69cce2b8a17fd40fa279b83b68c84"}]},"branch":"refs/heads/master"},"de9603b234d3cd3a0e79eb6f32627433c0e97543":{"kind":"REWORK","_number":3,"created":"2019-12-23 00:01:05.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/85/700085/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/85/700085/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/3"}}},"commit":{"parents":[{"commit":"77bc7d77dad5d95c6e87ba086de3d6e94f0014ac","subject":"Add Debian support for initialising DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/77bc7d77dad5d95c6e87ba086de3d6e94f0014ac"}]}],"author":{"name":"Jake Yip","email":"jake.yip@unimelb.edu.au","date":"2019-12-18 06:23:55.000000000","tz":660},"committer":{"name":"Jake Yip","email":"jake.yip@unimelb.edu.au","date":"2019-12-22 23:58:34.000000000","tz":660},"subject":"Add clustering support for ovsdb","message":"Add clustering support for ovsdb\n\nChange-Id: I740c3cf6ac907a0d6854680ee89b7ca60f3407d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/de9603b234d3cd3a0e79eb6f32627433c0e97543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/de9603b234d3cd3a0e79eb6f32627433c0e97543"}]},"branch":"refs/heads/master"},"82cadd3a280e60e12f6720b3842b9e1ffbd7c9b2":{"kind":"REWORK","_number":4,"created":"2020-01-07 05:27:45.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/85/700085/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/85/700085/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/85/700085/4"}}},"commit":{"parents":[{"commit":"77bc7d77dad5d95c6e87ba086de3d6e94f0014ac","subject":"Add Debian support for initialising DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/77bc7d77dad5d95c6e87ba086de3d6e94f0014ac"}]}],"author":{"name":"Jake Yip","email":"jake.yip@unimelb.edu.au","date":"2019-12-18 06:23:55.000000000","tz":660},"committer":{"name":"Jake Yip","email":"jake.yip@unimelb.edu.au","date":"2020-01-07 05:27:25.000000000","tz":660},"subject":"Add clustering support for ovsdb","message":"Add clustering support for ovsdb\n\nChange-Id: I740c3cf6ac907a0d6854680ee89b7ca60f3407d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/82cadd3a280e60e12f6720b3842b9e1ffbd7c9b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/82cadd3a280e60e12f6720b3842b9e1ffbd7c9b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
