)]}'
{"id":"openstack%2Fpuppet-ovn~743896","triplet_id":"openstack%2Fpuppet-ovn~master~I04e4fff5d37207bb19b72527dbf901cce72644a1","project":"openstack/puppet-ovn","branch":"master","topic":"prep-release-victoria-m2","hashtags":[],"change_id":"I04e4fff5d37207bb19b72527dbf901cce72644a1","subject":"Prepare Victoria M2","status":"MERGED","created":"2020-07-30 05:36:46.000000000","updated":"2020-07-30 14:07:09.000000000","submitted":"2020-07-30 14:07:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743896-1596118029774-689b57e4","meta_rev_id":"bd98acde3d48a1c2517d8da24026d4355a91580c","_number":743896,"virtual_id_number":743896,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-30 14:07:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-07-30 12:57:48.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-07-30 13:59:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-07-30 13:59:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-30 12:57:48.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-07-30 13:59:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-07-30 14:07:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9606743b495a5e26961dbce3b7ca69f28aac5528","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-07-30 05:36:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6532831198d5b47f4a5c4123f52ef057d419ca67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 05:47:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/0e65d13d8f5a4160b6b9f1df834c588d : SUCCESS in 2m 03s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f0d0b94dd06434e84210d8823037e86 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"40b5fc67e4fb68a7f9ebc43ea31d2a5c41facf72","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-30 12:57:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"99773b58019baaa78be1495edb245230cb75f062","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-07-30 13:59:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7ca3dea56c3730003d6edfb6a34ba8c99da9528","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 13:59:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e482c6a36166aac74c5bbf3f713e2ee3a4f7a5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 14:07:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4abb1f9742567851f9b6430904b4cf0d5290910d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 14:07:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/f22bb7daded540869b4db3a034cb71e1 : SUCCESS in 2m 40s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e76be8e5032443093d262bc2b963868 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"533055a3618734f45fec672f5d8f1f9c0347e417","revisions":{"533055a3618734f45fec672f5d8f1f9c0347e417":{"kind":"REWORK","_number":1,"created":"2020-07-30 05:36:46.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/96/743896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/96/743896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/96/743896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/96/743896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/96/743896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/96/743896/1"}}},"commit":{"parents":[{"commit":"f64067083b7f145c5717b3778ad1837ee59592b6","subject":"Fix beaker-rspec url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/f64067083b7f145c5717b3778ad1837ee59592b6"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-07-30 05:35:29.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-07-30 05:35:29.000000000","tz":480},"subject":"Prepare Victoria M2","message":"Prepare Victoria M2\n\nUpdate the version metadata for Victoria milestone M2\n\nChange-Id: I04e4fff5d37207bb19b72527dbf901cce72644a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/533055a3618734f45fec672f5d8f1f9c0347e417"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/533055a3618734f45fec672f5d8f1f9c0347e417"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
