)]}'
{"id":"openstack%2Fpuppet-ovn~779804","triplet_id":"openstack%2Fpuppet-ovn~master~I4b402c58cc3c39cd2650d91d12d607fe0171e1d0","project":"openstack/puppet-ovn","branch":"master","topic":"hw-offload","hashtags":[],"change_id":"I4b402c58cc3c39cd2650d91d12d607fe0171e1d0","subject":"Deprecate duplicated implementation for hw-offload option","status":"MERGED","created":"2021-03-10 15:14:38.000000000","updated":"2021-03-26 07:42:28.000000000","submitted":"2021-03-26 07:41:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779804-hw-offload","meta_rev_id":"febaa316d231f4d96a99999268a11a83eecf0426","_number":779804,"virtual_id_number":779804,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 07:41:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-03-26 07:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-03-12 15:14:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-03-26 07:12:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-10 15:37:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-12 15:14:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-03-26 07:12:26.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"34465e5438efdbcebc6521fb5ed8d0e75afd0673","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-10 15:14:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd32590cede262c1f596c96ff1f95d1a7df739f0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-10 15:15:38.000000000","message":"Topic ovn-distributed-vlan-traffic removed","accounts_in_message":[],"_revision_number":1},{"id":"4a5fc2548c8f41536fda157e9223ef68c8d2783c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-10 15:15:52.000000000","message":"Topic set to hw-offload","accounts_in_message":[],"_revision_number":1},{"id":"7c3729aee74512bd8a76d9de010b5e1f6d5b63ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 15:37:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3371cbba0c284ae1ba0a375263c04ef9 : FAILURE in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f60217fa2f5495c86eb3a4ce19183f8 : SUCCESS in 4m 18s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e348f26ccfdd4ae5bbfe8f38cfa4c54d : SUCCESS in 8m 12s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/4767c814a2c04fb7b047f674f57efc52 : SUCCESS in 9m 03s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e7d9e20f68b4875986dea1d19cd6704 : SUCCESS in 6m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23df6022d9f9464b907249f2d72bc9a7 : SUCCESS in 6m 17s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/6d9814684d1542c9ab7d90008a9be002 : SUCCESS in 12m 15s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/952591a69a504c19a7951f6c6b11afd6 : SUCCESS in 12m 59s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03255c256df74ccaaf728bc3aa14355e : SUCCESS in 8m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d83c9c8b548430b829bb020e2511421 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6a8f598fd9cc037d89bac3f8920952d462346028","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-10 15:47:31.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ce427742192a8e230fdeb9724df234000a2d06c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 16:37:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/232f20ec93074a2bb513b06eb3670a50 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08477148a3204a269ed73d0e7e19dae9 : SUCCESS in 4m 28s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1dcb9f6a805e4f038f388e69815801c4 : SUCCESS in 10m 14s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/39ff94f00ae84703852b7056c78858fe : SUCCESS in 10m 10s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25194fafcdcd4513bb7e4af68694701e : SUCCESS in 9m 03s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4e70adee84549e0bfc28850fae9eea1 : SUCCESS in 7m 52s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/eb238f923cbc4f43bb33cceafc26e7f8 : SUCCESS in 14m 21s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a53bf135b744b56a01207dd9fe35696 : SUCCESS in 13m 34s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35fffcb28db24512ab0440b741871209 : SUCCESS in 8m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee8cd563d3074e0385e44845dba01ada : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":2},{"id":"51a54139aa4f7103e380764f2ddcf00e4fddf1b5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-12 15:14:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c444e0afe1cb020e44c5022fd93c902769dee8fd","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-03-26 07:12:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cd0d743d17f0022feb8601e86b89fe1cd903a7b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 07:12:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"051d624274da264552968acaf36b3301002515bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 07:41:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3055ffceda624f9ea8cf9b91551c5921 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44ea9939b3f940d38633d6c0d7103328 : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/195fdbaa0b4541b59c9b33c05bd7fcdc : SUCCESS in 8m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9d1e8a3a993424dbe94ff1b9edef019 : SUCCESS in 6m 18s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/5330647e97e24da5a671b4df9a082d81 : SUCCESS in 16m 35s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0de473c84106450abc298949a1446258 : SUCCESS in 7m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e68593559f6846a8a1ad8f5e7fba8b28 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"2b470882c0773ac05312587361bfb49669b02b77","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 07:41:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"febaa316d231f4d96a99999268a11a83eecf0426","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 07:42:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17487db7eb6241049aa20b2c3f6d0cb8 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c60fdb9540bc5dfc60cf05d56ad9d8fe45164343","revisions":{"a90649a5d75b7a0fa22e959c0c8850157c1da3f5":{"kind":"REWORK","_number":1,"created":"2021-03-10 15:14:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/779804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/04/779804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/04/779804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/04/779804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/04/779804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/04/779804/1"}}},"commit":{"parents":[{"commit":"73a1d569220d2601d9446838caaacea4168d5ac4","subject":"Set external_ids:ovn-chassis-mac-mappings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/73a1d569220d2601d9446838caaacea4168d5ac4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-10 15:10:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-10 15:14:11.000000000","tz":540},"subject":"Deprecate duplicated implementation for hw-offload option","message":"Deprecate duplicated implementation for hw-offload option\n\nCurrently the hw-offload option is managed by two modules,\npuppet-vswitch and puppet-ovn. Because this option is generic and\nvalid for both ovs and ovn this change deprecate the parameter from\npuppet-ovn.\n\nDepends-on: https://review.opendev.org/779802\nChange-Id: I4b402c58cc3c39cd2650d91d12d607fe0171e1d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/a90649a5d75b7a0fa22e959c0c8850157c1da3f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/a90649a5d75b7a0fa22e959c0c8850157c1da3f5"}]},"branch":"refs/heads/master"},"c60fdb9540bc5dfc60cf05d56ad9d8fe45164343":{"kind":"REWORK","_number":2,"created":"2021-03-10 15:47:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/779804/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/04/779804/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/04/779804/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/04/779804/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/04/779804/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/04/779804/2"}}},"commit":{"parents":[{"commit":"73a1d569220d2601d9446838caaacea4168d5ac4","subject":"Set external_ids:ovn-chassis-mac-mappings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/73a1d569220d2601d9446838caaacea4168d5ac4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-10 15:10:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-10 15:47:19.000000000","tz":0},"subject":"Deprecate duplicated implementation for hw-offload option","message":"Deprecate duplicated implementation for hw-offload option\n\nCurrently the hw-offload option is managed by two modules,\npuppet-vswitch and puppet-ovn. Because this option is generic and\nvalid for both ovs and ovn this change deprecate the parameter from\npuppet-ovn.\n\nDepends-on: https://review.opendev.org/779802\nChange-Id: I4b402c58cc3c39cd2650d91d12d607fe0171e1d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/c60fdb9540bc5dfc60cf05d56ad9d8fe45164343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/c60fdb9540bc5dfc60cf05d56ad9d8fe45164343"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
