)]}'
{"id":"openstack%2Fpuppet-ovn~801455","triplet_id":"openstack%2Fpuppet-ovn~master~Ia78c0873e7ac6488b3048ae8475cc87ec40022bb","project":"openstack/puppet-ovn","branch":"master","topic":"ovn-monitor-all","hashtags":[],"change_id":"Ia78c0873e7ac6488b3048ae8475cc87ec40022bb","subject":"Add ovn-monitor-all option to ovn-controller","status":"MERGED","created":"2021-07-20 12:26:17.000000000","updated":"2021-07-22 15:07:54.000000000","submitted":"2021-07-22 15:06:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801455-ovn-monitor-all","meta_rev_id":"76e25667a9de89554a095b4142c4d00f3f03f156","_number":801455,"virtual_id_number":801455,"owner":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-22 15:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-07-22 14:14:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-07-22 14:14:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-20 12:47:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-21 21:46:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-07-22 08:35:39.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"bcab9f63371d2be19012dd82faa99196adf23b68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2021-07-20 12:26:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c6c0ce116f2f22bf78b7c26b10a08c2beddaa21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 12:47:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8750e32cc3c4ee88c48747dbe6c9669 : FAILURE in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fddfd917f5b3495bb7968f6109404e26 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbe58a5396b74673b4390f0709680a36 : SUCCESS in 3m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/296d67451e8844c4a0e040f9176de025 : SUCCESS in 10m 05s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/32a478fbb0ad406c93983f0d446701ae : SUCCESS in 10m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cae6bebc91264d57a30b7d387e308dbc : SUCCESS in 7m 24s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9694537c78a0402a9418ce8e9eac5430 : SUCCESS in 6m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/948c1342400249f89a028717a6083493 : SUCCESS in 7m 09s (non-voting)\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6818dc88a5740669cdd1bdbee60aefe : SUCCESS in 16m 44s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68e4ee056e4a4906a097473c6046efaf : SUCCESS in 7m 50s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/58a0a934e2634d1b9b67f8d9e41fbc67 : SUCCESS in 16m 38s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc9019fe81d8420883c8683d42bd552d : SUCCESS in 8m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/509fe6755f664433b9843dbd541307f6 : SUCCESS in 7m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"470481d077fd3fc567470c0935e340fc70e50a3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2021-07-21 13:59:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f4c2fc55f897670037674d1ffb4328bd7970d45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 14:28:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f3394a7b7f547cfb5ed61feae08fe55 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31a6151d2e554c66838af24cc3caa41c : SUCCESS in 3m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1774e1df07214e0ebf13a7ab1b0f3212 : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8fb5318a40547c5ac1524d6e436fdbc : SUCCESS in 10m 33s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/342aac556cc3409791a82d16864ec3f5 : SUCCESS in 10m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26f43d46769a415f80974aa2c5d083eb : SUCCESS in 7m 27s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d179426ff747455098509023fe85f2df : SUCCESS in 7m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e943b1bbb8804094891f0a1e3060c298 : SUCCESS in 6m 35s (non-voting)\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec5e08ade5014d8f85cb41d6c3e66d98 : SUCCESS in 17m 35s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53ea6aa265cb43dcbe1229b400fa6e6b : SUCCESS in 8m 06s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/f7d8e48d80d641569fd26596336dfbd4 : SUCCESS in 17m 34s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23599d399e7f47f2a4680bfad93fed38 : SUCCESS in 7m 57s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/182384f9f8534790b2cd11b96054450b : SUCCESS in 8m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b5db0141e6a2392ff750c5a0933c6fc73964fab7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-21 21:46:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7a0f88d1238f00e90dffbc710862fc6841e0e512","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 12:43:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a02790336b7b0a6eb4c1f9226916b28ec3532142","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 12:44:29.000000000","message":"Patch Set 2: Code-Review-1 -Workflow\n\nSorry I missed one thing. Please add a release note about the new parameter.","accounts_in_message":[],"_revision_number":2},{"id":"b3ee84903961537a5d875bac606691a0a3d8c6d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 12:44:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3593468bb0f03cc11c07e21c54770b2e4c573753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 13:07:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50e33ce9d7c540aa9e506390c18036da : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b519c3eee444bf5b69d5a10e9f3d205 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8330c45d5dbe4f089acf7051572be018 : SUCCESS in 3m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1d3ce14d3224290837f178a29c730fa : SUCCESS in 11m 02s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/2bd2086abc774a889aaaa9279aaebba9 : SUCCESS in 9m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5c434c8972d413fad7fc6d70cc69026 : SUCCESS in 7m 16s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6016d0dda114a758a7343b811293456 : SUCCESS in 6m 57s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ace7287b65f48c18f20347df8a3abc2 : SUCCESS in 18m 29s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc40317e254848079162b5a486dc3631 : SUCCESS in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"723ee97f24bb1578b06ef3625a936f01978a0112","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 14:13:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eb2150e3f7bed6e0318ba44b309347cbfd1f89e9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 14:14:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nA release note file is added. The implementation was already approved so I\u0027ll merge this change now.","accounts_in_message":[],"_revision_number":3},{"id":"6cee543307b9c660312796d8c2cb1212be7a2d10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 14:39:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caadcc359f1446678343a64f5f6bee5c : SUCCESS in 4m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0f8aabce1c94e7aab067fb93f4668ca : SUCCESS in 4m 24s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45e4c546d57c4ce2a130df007896c751 : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fbfe182714f9406a87639c939a583f60 : SUCCESS in 10m 53s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/28e0aede09844b10ada275f76187c69e : SUCCESS in 10m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c6d1455010745a1ad53274fe99a9835 : SUCCESS in 8m 22s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1636baf047554998b54a13b4fd4b9bd1 : SUCCESS in 7m 30s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cd702061d674338a34ae14c45504fd7 : SUCCESS in 10m 34s (non-voting)\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/53f3eefc782a4cc2981b7f011b71f7e6 : SUCCESS in 19m 24s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9329efaa8980473ebcf1690bd79f0d4d : SUCCESS in 11m 06s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/67822ecdb016467d90531833a9616fcb : SUCCESS in 21m 09s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/953d6376409a41aba8e6a9c880e37fe9 : SUCCESS in 11m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9eb357dbbe6541a3b7b3e941c9916074 : SUCCESS in 11m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7818000dc9e48569f3a512d2ead480c : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":3},{"id":"27bb2d136d33aabeb6c7c8e204940dfc0343bf26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 14:40:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5d54fce16be8da9acd0f1440d67ddb7ae7b92308","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 15:06:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8f900c97cf54803b8dd876ab453d8fd : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5668800835c94728994d8ad91e00940d : SUCCESS in 3m 52s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9134b512ad9845daad5127e3a164595b : SUCCESS in 3m 56s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/de65673755ca48dcbb6f2d9de378362f : SUCCESS in 10m 47s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/93dc789b325240fbb20377bd771f6948 : SUCCESS in 10m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc71d2e556f447ea985bcac65806835f : SUCCESS in 7m 15s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1be78de17f264ba0a312b8f981823dad : SUCCESS in 7m 10s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e575629b508f499ca9e27c267e6084c6 : SUCCESS in 19m 03s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ceadd37356d34f41accf6649f923e590 : SUCCESS in 8m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3df35a3dd483419ba0acc124c96816bc : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":3},{"id":"46db9c6b0e227c8379ea2d2783097d71bb5b65fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 15:06:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"76e25667a9de89554a095b4142c4d00f3f03f156","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 15:07:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e84878b2f5b4bdb891e3204c8bce080 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7a1bb545198d8c864c619b30f4bc13417d764bbd","revisions":{"5fe02997187f11ad30878684741a4eb47f3bb661":{"kind":"REWORK","_number":1,"created":"2021-07-20 12:26:17.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/55/801455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/55/801455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/55/801455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/55/801455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/55/801455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/55/801455/1"}}},"commit":{"parents":[{"commit":"6b085474cf1fd9fc9d2f3c0d2c9143250028bd98","subject":"Merge \"Drop Fedora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/6b085474cf1fd9fc9d2f3c0d2c9143250028bd98"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2021-07-20 12:08:49.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2021-07-20 12:26:15.000000000","tz":120},"subject":"Add ovn-monitor-all option to ovn-controller","message":"Add ovn-monitor-all option to ovn-controller\n\nWith monitor-all option, all ovn-controllers\ndo not enable conditional monitoring there\nby reducing the load on the Southbound ovsdb-server.\n\nPartial-Bug: #1936781\nSigned-off-by: Kamil Sambor \u003cksambor@redhat.com\u003e\nChange-Id: Ia78c0873e7ac6488b3048ae8475cc87ec40022bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/5fe02997187f11ad30878684741a4eb47f3bb661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/5fe02997187f11ad30878684741a4eb47f3bb661"}]},"branch":"refs/heads/master"},"e1338a97464d19320df3e2a7bc67642c29eaf16f":{"kind":"REWORK","_number":2,"created":"2021-07-21 13:59:03.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/55/801455/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/55/801455/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/55/801455/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/55/801455/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/55/801455/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/55/801455/2"}}},"commit":{"parents":[{"commit":"6b085474cf1fd9fc9d2f3c0d2c9143250028bd98","subject":"Merge \"Drop Fedora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/6b085474cf1fd9fc9d2f3c0d2c9143250028bd98"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2021-07-20 12:08:49.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2021-07-21 13:58:55.000000000","tz":120},"subject":"Add ovn-monitor-all option to ovn-controller","message":"Add ovn-monitor-all option to ovn-controller\n\nWith monitor-all option, all ovn-controllers\ndo not enable conditional monitoring there\nby reducing the load on the Southbound ovsdb-server.\n\nPartial-Bug: #1936781\nSigned-off-by: Kamil Sambor \u003cksambor@redhat.com\u003e\nChange-Id: Ia78c0873e7ac6488b3048ae8475cc87ec40022bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e1338a97464d19320df3e2a7bc67642c29eaf16f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e1338a97464d19320df3e2a7bc67642c29eaf16f"}]},"branch":"refs/heads/master"},"7a1bb545198d8c864c619b30f4bc13417d764bbd":{"kind":"REWORK","_number":3,"created":"2021-07-22 14:13:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/801455/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/55/801455/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/55/801455/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/55/801455/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/55/801455/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/55/801455/3"}}},"commit":{"parents":[{"commit":"6b085474cf1fd9fc9d2f3c0d2c9143250028bd98","subject":"Merge \"Drop Fedora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/6b085474cf1fd9fc9d2f3c0d2c9143250028bd98"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2021-07-20 12:08:49.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-22 14:13:22.000000000","tz":540},"subject":"Add ovn-monitor-all option to ovn-controller","message":"Add ovn-monitor-all option to ovn-controller\n\nWith monitor-all option, all ovn-controllers\ndo not enable conditional monitoring there\nby reducing the load on the Southbound ovsdb-server.\n\nPartial-Bug: #1936781\nSigned-off-by: Kamil Sambor \u003cksambor@redhat.com\u003e\nChange-Id: Ia78c0873e7ac6488b3048ae8475cc87ec40022bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/7a1bb545198d8c864c619b30f4bc13417d764bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/7a1bb545198d8c864c619b30f4bc13417d764bbd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
