)]}'
{"id":"openstack%2Fpuppet-ovn~852233","triplet_id":"openstack%2Fpuppet-ovn~master~Ia7e85756882606912e06a6742cd35517b568dffe","project":"openstack/puppet-ovn","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-12 08:04:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia7e85756882606912e06a6742cd35517b568dffe","subject":"Ensure ovn-bridge-datapath-type is removed by default","status":"MERGED","created":"2022-08-05 04:55:23.000000000","updated":"2022-08-12 08:04:55.000000000","submitted":"2022-08-12 08:04:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852233","meta_rev_id":"2dffc2ac75d95e204ce53f39c7cd8c826f47c6c6","_number":852233,"virtual_id_number":852233,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-12 08:04:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-08-08 15:20:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-08-12 06:40:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-08-12 06:40:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-05 05:16:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-08 15:20:34.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-08-12 06:40:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"8c9fb97cdc14606b4f55947c92e14a247e41004e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-05 04:55:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a351d7427afc339b393cdaa51d2b09e96fb8506","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 05:16:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e26f4fde09e4fbf93bcee4fa19ded20\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9eb8b48fb51145a0bebd0652785fa1a8 : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6479b6947746420cb200c71176fd372e : FAILURE in 3m 56s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/753b44fc3d72418d91da32c2e45d2c66 : FAILURE in 3m 47s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4a74d8976e047678ac31720f89f152f : FAILURE in 17m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70603cc82b3d4da7a6e6b0f5868c960f : FAILURE in 10m 53s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e53b81df1be84e469f7ca5d6cb6ca662 : FAILURE in 10m 35s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b084aa5e6604bd6935d81f066270f7f : FAILURE in 10m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45b3320f6a344ae5956d64789559d463 : FAILURE in 11m 23s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f303599a5724dbcaf746d1612ae0ddd : FAILURE in 18m 00s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c7c2a645eb9423fbc7849b23349e86a : FAILURE in 9m 10s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af4eb4136f2d49968b58a7b1d03ca7a3 : FAILURE in 16m 59s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31163ea650234e45873b3f505cf7c16c : FAILURE in 12m 29s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/07235ff9a139412bb7f743ec94d64a4c : FAILURE in 18m 52s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba796ab9356d43d391e0c204b515b87c : FAILURE in 12m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0704b9a938704d0a9859d10c363aa6f5 : FAILURE in 12m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b762f11f6c3853c70b27eb8eba80381fbfd3153","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-05 07:13:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"da07f4baafe8d60cbbfc175ccca3acaa585430d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-05 07:26:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b4cd5392179be9bce0cf306e09bb4fd70f18b932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 07:50:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4575a687d8684d0c86a79837315f3178\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0eae064b79de44d4b39ebdbcde908cab : SUCCESS in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/565ccee2dedb4403aac6c880f3219a2b : FAILURE in 3m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23831916ea50493a90bcf7e945c30930 : FAILURE in 3m 26s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/696b02e36cc64023bc87bdfff2b6a473 : FAILURE in 16m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b7df2d19adb479fbfbe4b7918507ab6 : FAILURE in 11m 15s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09245a68b7b8473e8a65f142c889d522 : FAILURE in 9m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71fe7a63621146cea67282b54e2d8e3c : FAILURE in 9m 43s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1a16252e4e345ce929c422ef88d241c : FAILURE in 11m 07s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/228f3f85fa724e059146b366423f6809 : FAILURE in 15m 44s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cf6dcb79d274585967c7e0b56306ed5 : FAILURE in 12m 43s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a37b88aff2474039ab2150b5cbe97779 : FAILURE in 18m 55s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d125770f4144e17a90c710277b5b39a : FAILURE in 13m 27s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf8763b72b704d23bc001e855131bd01 : FAILURE in 22m 38s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c486b1a32a940f98bb8e0e0d8fb82b0 : FAILURE in 10m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/440a5cfbae3f4b48a62ef563cb9a5b06 : FAILURE in 11m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a5e19340ff85658ae408b37c3486694cde161b42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-05 07:55:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ea5d839c5a3cc252fd89e9dfe2fc56facf13024d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 09:00:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6d47e5613d2401aa806a80fa38e14f6\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be614f08a2c94e139fa76fe20d0030da : SUCCESS in 3m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96710d65e08f41db88bbfb4831ff7173 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a328927f7a2f44a0894d9fbeb73ab2ab : SUCCESS in 3m 23s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a530f20c2ff7482b9e83e44dcc0e3836 : SUCCESS in 12m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a33418f1e59424e9c92783cd17601f8 : SUCCESS in 8m 51s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25a104842133415dbc4b981e7ef7409d : SUCCESS in 8m 47s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4c53ea3e8384631bb841822aeb5806f : SUCCESS in 8m 40s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b6be30190be48cab3d179eb964caf30 : FAILURE in 35m 34s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e628deededbc4baf9835373cbd07c69a : SUCCESS in 1h 02m 26s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b14a687968c94b86868243a7b9dc6bea : FAILURE in 28m 25s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a30300eaaeda4aacb1faba72f5262a64 : SUCCESS in 50m 27s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c320d3a8737d4b3a83f3cdfd2c10dd00 : SUCCESS in 10m 48s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/96e00b0b092c4c9ba44cd2c0b4d2ff24 : SUCCESS in 17m 34s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89f838a184d64def9b285f261a98eea2 : SUCCESS in 10m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/100ecbbaa8ac40d39a1913f37fd79641 : SUCCESS in 10m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"05c27d4f7daf1874558ba19057da2aac2efdb683","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-08 15:20:34.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"79a6a97f2bf6769cbffde693730e7db1c5b4615c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-12 06:40:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ed8b0681c4ef83c4b8d54996bfb19a04e4939313","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 06:41:59.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"62422d7b4491dcb2dc39fe2bbe4fb1ee8d01937e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 08:04:55.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36cbda19aa3c47338eb691b7de03eb03\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/504ecb18895f42eebd3a6090e92df6b3 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1e0851140bf41b59ca8e9fcb84e1f24 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd3498abbc0d488bb9248e7c1b983c1b : SUCCESS in 3m 52s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7b8ab1ff06f4b1fbef96dcf5faf2864 : SUCCESS in 10m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98dd42de892a4aa689f07b175c1358aa : SUCCESS in 11m 01s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c521791db444882847ae03a4318fc45 : SUCCESS in 5m 48s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/855ada50ec454657af431d88ad3d1608 : SUCCESS in 1h 21m 12s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7ac581606574b05a847cf6c36179274 : SUCCESS in 46m 57s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c12fc9ed8c37491ba8d555436f2c3333 : SUCCESS in 23m 26s","accounts_in_message":[],"_revision_number":4},{"id":"2dffc2ac75d95e204ce53f39c7cd8c826f47c6c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 08:04:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"59ffa355e9e1e4f391cdfa87d6bc0c3d7e722dd6","revisions":{"2d3ae27f2b2827dc73782363427c3f729f97b39f":{"kind":"REWORK","_number":1,"created":"2022-08-05 04:55:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/852233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/33/852233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/1"}}},"commit":{"parents":[{"commit":"42a579c75ee89522dd33c9f63c9e3f89bd670639","subject":"Merge \"Fix logics to assert ovn::controller::(bridge|port)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/42a579c75ee89522dd33c9f63c9e3f89bd670639"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-05 04:54:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-05 04:54:12.000000000","tz":540},"subject":"Ensure ovn-bridge-datapath-type is removed by default","message":"Ensure ovn-bridge-datapath-type is removed by default\n\nThis is follow-up of 5c809de01630ee7eede500106143810ca57b58b5 and\nensures the ovn-bridge-datapath-type option is removed by default.\nCurrent Puppet does not ensure the value in case a nil value is passed\nthus setting value \u003d\u003e undef has no effect.\n\nChange-Id: Ia7e85756882606912e06a6742cd35517b568dffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/2d3ae27f2b2827dc73782363427c3f729f97b39f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/2d3ae27f2b2827dc73782363427c3f729f97b39f"}]},"branch":"refs/heads/master"},"7336991130f438463a867dd7dd5e7044f0c08dd1":{"kind":"REWORK","_number":2,"created":"2022-08-05 07:13:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/852233/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/33/852233/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/2"}}},"commit":{"parents":[{"commit":"4f58f3ccd36f46dc20556653d5304ec1fe9c3865","subject":"Merge \"Use native property to set mac table size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/4f58f3ccd36f46dc20556653d5304ec1fe9c3865"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-05 04:54:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-05 07:13:42.000000000","tz":540},"subject":"Ensure ovn-bridge-datapath-type is removed by default","message":"Ensure ovn-bridge-datapath-type is removed by default\n\nThis is follow-up of 5c809de01630ee7eede500106143810ca57b58b5 and\nensures the ovn-bridge-datapath-type option is removed by default.\nCurrent Puppet does not ensure the value in case a nil value is passed\nthus setting value \u003d\u003e undef has no effect.\n\nChange-Id: Ia7e85756882606912e06a6742cd35517b568dffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/7336991130f438463a867dd7dd5e7044f0c08dd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/7336991130f438463a867dd7dd5e7044f0c08dd1"}]},"branch":"refs/heads/master"},"d2fdf42887d8502dcb833651566b0f4159517c1c":{"kind":"REWORK","_number":3,"created":"2022-08-05 07:26:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/852233/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/33/852233/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/3"}}},"commit":{"parents":[{"commit":"4f58f3ccd36f46dc20556653d5304ec1fe9c3865","subject":"Merge \"Use native property to set mac table size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/4f58f3ccd36f46dc20556653d5304ec1fe9c3865"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-05 04:54:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-05 07:25:59.000000000","tz":540},"subject":"Ensure ovn-bridge-datapath-type is removed by default","message":"Ensure ovn-bridge-datapath-type is removed by default\n\nThis is follow-up of 5c809de01630ee7eede500106143810ca57b58b5 and\nensures the ovn-bridge-datapath-type option is removed by default.\nCurrent Puppet does not ensure the value in case a nil value is passed\nthus setting value \u003d\u003e undef has no effect.\n\nChange-Id: Ia7e85756882606912e06a6742cd35517b568dffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/d2fdf42887d8502dcb833651566b0f4159517c1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/d2fdf42887d8502dcb833651566b0f4159517c1c"}]},"branch":"refs/heads/master"},"59ffa355e9e1e4f391cdfa87d6bc0c3d7e722dd6":{"kind":"REWORK","_number":4,"created":"2022-08-05 07:55:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/852233/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/33/852233/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/33/852233/4"}}},"commit":{"parents":[{"commit":"4f58f3ccd36f46dc20556653d5304ec1fe9c3865","subject":"Merge \"Use native property to set mac table size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/4f58f3ccd36f46dc20556653d5304ec1fe9c3865"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-05 04:54:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-05 07:55:41.000000000","tz":540},"subject":"Ensure ovn-bridge-datapath-type is removed by default","message":"Ensure ovn-bridge-datapath-type is removed by default\n\nThis is follow-up of 5c809de01630ee7eede500106143810ca57b58b5 and\nensures the ovn-bridge-datapath-type option is removed by default.\nCurrent Puppet does not ensure the value in case a nil value is passed\nthus setting value \u003d\u003e undef has no effect.\n\nChange-Id: Ia7e85756882606912e06a6742cd35517b568dffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/59ffa355e9e1e4f391cdfa87d6bc0c3d7e722dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/59ffa355e9e1e4f391cdfa87d6bc0c3d7e722dd6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
