)]}'
{"id":"openstack%2Fpuppet-ovn~854975","triplet_id":"openstack%2Fpuppet-ovn~master~I8825a7438ba20be97da71acabd57604e7948e130","project":"openstack/puppet-ovn","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-29 16:36:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8825a7438ba20be97da71acabd57604e7948e130","subject":"Remove redundant resource order","status":"MERGED","created":"2022-08-29 09:21:18.000000000","updated":"2022-08-29 16:36:36.000000000","submitted":"2022-08-29 16:36:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854975","meta_rev_id":"fe2eca1df97e54a0847983c6bd6c833ebd76b47b","_number":854975,"virtual_id_number":854975,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-29 16:36:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-08-29 10:57:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-08-29 15:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-08-29 15:12:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-29 10:27:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-29 10:57:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-08-29 15:12:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"47af7411d80952f3b2176644efe1ef0bf1cc173b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-29 09:21:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13baf4354ceb4e65b133d311aff51e9c8e146753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 10:27:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/139080314a374bfe8d886c7e3cdb1a6c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f3a0a90ccb947e58cefcc7ccb6cd2d8 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69c662fee0a44d29972274333c80a3dd : SUCCESS in 3m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b1c9e72c00b46098a5cb1581af4539b : SUCCESS in 4m 00s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/feb8356a782a4a69b176a2b895fb1764 : SUCCESS in 16m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/864ec46135d54ea59b49ee52f0d3df29 : SUCCESS in 10m 45s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1d64fee6e5f44bfbed949df3336a273 : SUCCESS in 10m 31s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40d385aff07143deb5fb6723c719fb1b : SUCCESS in 10m 06s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/466ea758b70a49b5a3773c2c45c5655a : FAILURE in 9m 19s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/460e3637299a485b974b9f18c96a2906 : SUCCESS in 1h 03m 30s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa4caba8ff2b42eb947732ae29cd66db : FAILURE in 10m 16s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a895cdef6a945d5aae2e8d2fa5f90d4 : SUCCESS in 53m 03s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0135b432614b45828b79dd19d192709c : SUCCESS in 9m 42s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d412ed11182b462e920d3e240bc5fa72 : SUCCESS in 16m 53s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f3faf0f80d0451aa88f763ccdf13ef6 : SUCCESS in 9m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/015fe842205c49ac80c845b4f9746136 : SUCCESS in 11m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d60d01c2347e9c08f3e350b8a61008de5c55c733","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-29 10:57:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2479582dff139777dbcc1931f1e38a0849c63321","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-29 15:12:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8173d1c831a4d95b2f7f9163c0c5a52fa4e42b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:16:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe9c44b4827a11711b5f0afaa524bba4c1065bed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 16:36:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03c53ad4dff6474c965f7acdf4be0ff8\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bec9dd84d3d64886a90c9af1b8138968 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9415f98c6dd64372afb8bc839c3c1f70 : SUCCESS in 3m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccc291ac9d494f4bb95a553c742476a7 : SUCCESS in 4m 26s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/35f5389a8f7645b4a50449a60f4c1b46 : SUCCESS in 17m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7365bea2a28490e8ad0f846051b5383 : SUCCESS in 10m 45s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9fc084d55f14c4d97c355ef3413ead1 : SUCCESS in 10m 49s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7fabd872b55a47b49f778ec8e555ded1 : SUCCESS in 1h 10m 46s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/91ce5a3e4f614932bfec18cbb0e821e5 : SUCCESS in 56m 50s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3515e9d4aa3a428da75bfae931f57ec2 : SUCCESS in 15m 45s","accounts_in_message":[],"_revision_number":1},{"id":"fe2eca1df97e54a0847983c6bd6c833ebd76b47b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 16:36:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a377b3649499532a8b59df5d2334bd6c28dc538b","revisions":{"a377b3649499532a8b59df5d2334bd6c28dc538b":{"kind":"REWORK","_number":1,"created":"2022-08-29 09:21:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/854975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/75/854975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/75/854975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/75/854975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/75/854975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/75/854975/1"}}},"commit":{"parents":[{"commit":"3055bbc5aa312fa92c9eb39f33e54825e805a612","subject":"Test default behavior of ovn::controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/3055bbc5aa312fa92c9eb39f33e54825e805a612"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-29 09:19:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-29 09:20:50.000000000","tz":540},"subject":"Remove redundant resource order","message":"Remove redundant resource order\n\nThe puppet-vswitch module already makes sure that the vs_config\nresource automatically requires the openvswitch service. Thus defining\nthe same order is not necessary.\n\nChange-Id: I8825a7438ba20be97da71acabd57604e7948e130\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/a377b3649499532a8b59df5d2334bd6c28dc538b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/a377b3649499532a8b59df5d2334bd6c28dc538b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
